)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~654384","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~I8eb7eb4f7a9ceed21401318e3395fe4450db13b6","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I8eb7eb4f7a9ceed21401318e3395fe4450db13b6","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:16:46.000000000","updated":"2019-04-23 02:10:34.000000000","submitted":"2019-04-23 02:10:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654384-1555985434339-11ac1cc1","meta_rev_id":"0a5716d340658acd602b64038b9508c4197689ec","_number":654384,"virtual_id_number":654384,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"date":"2019-04-22 23:03:34.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-23 02:10:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 10:53:38.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 10:53:38.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 10:53:38.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 23:03:34.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-23 02:10:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"462ef4c588c3238d467059a8564d0465e0952d48","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:16:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8536970f12adee0398666e2cf7f7c874ca935240","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:47:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/84/654384/1/check/openstack-ansible-linters/975d895/ : SUCCESS in 9m 09s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/84/654384/1/check/openstack-ansible-functional-debian-stable/c09d2c9/ : SUCCESS in 1h 05m 04s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/84/654384/1/check/openstack-ansible-functional-centos-7/91e9825/ : SUCCESS in 56m 02s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/84/654384/1/check/openstack-ansible-functional-ubuntu-bionic/f63d06b/ : SUCCESS in 46m 15s\n- openstack-tox-docs http://logs.openstack.org/84/654384/1/check/openstack-tox-docs/72eb5a6/html/ : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"57fccb9cab7ab159440dce820a9ceef2efd94358","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 10:53:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7595e94e973a1ecef28b15b7579ae258232e6157","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 10:53:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"359cb17b5c1f78712a3dd2908d4dc10496fbbce3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:06:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/84/654384/1/gate/openstack-ansible-linters/913e3e9/ : SUCCESS in 6m 00s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/84/654384/1/gate/openstack-ansible-functional-debian-stable/80d2698/ : SUCCESS in 59m 45s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/84/654384/1/gate/openstack-ansible-functional-centos-7/68c1067/ : FAILURE in 48m 29s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/84/654384/1/gate/openstack-ansible-functional-ubuntu-bionic/c26873b/ : SUCCESS in 46m 22s\n- openstack-tox-docs http://logs.openstack.org/84/654384/1/gate/openstack-tox-docs/2021d17/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0bd9d61a704cfa751d6ea3a847c63dcd46a7e632","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-22 23:03:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d8ee2df661ce9cccecfaca9b234d28d045da8f34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 00:17:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/84/654384/1/check/openstack-ansible-linters/976a146/ : SUCCESS in 5m 34s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/84/654384/1/check/openstack-ansible-functional-debian-stable/7ac4ce6/ : SUCCESS in 57m 45s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/84/654384/1/check/openstack-ansible-functional-centos-7/f3e5cc1/ : SUCCESS in 59m 33s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/84/654384/1/check/openstack-ansible-functional-ubuntu-bionic/474549e/ : SUCCESS in 46m 55s\n- openstack-tox-docs http://logs.openstack.org/84/654384/1/check/openstack-tox-docs/42390c2/html/ : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7d4933ac603b1c595051d1fafd831f4c2639ca52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 00:17:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a50b61e679dc6ee8e74549d085eedd9da9ee7609","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 02:10:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/84/654384/1/gate/openstack-ansible-linters/e233b30/ : SUCCESS in 4m 57s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/84/654384/1/gate/openstack-ansible-functional-debian-stable/0e8caca/ : SUCCESS in 46m 18s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/84/654384/1/gate/openstack-ansible-functional-centos-7/a1691eb/ : SUCCESS in 1h 43m 45s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/84/654384/1/gate/openstack-ansible-functional-ubuntu-bionic/2333c3d/ : SUCCESS in 46m 02s\n- openstack-tox-docs http://logs.openstack.org/84/654384/1/gate/openstack-tox-docs/baa0258/html/ : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3533813b763ddc3c857d32b93e0698ed06ff6058","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 02:10:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41ac6d0509f292236f7253b2c8ba33440fb14a9e","revisions":{"41ac6d0509f292236f7253b2c8ba33440fb14a9e":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:16:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/654384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/84/654384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/84/654384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/84/654384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/84/654384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/84/654384/1"}}},"commit":{"parents":[{"commit":"070f0d48be8cd81e690379eb86b3151dd29786eb","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/070f0d48be8cd81e690379eb86b3151dd29786eb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:16:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:16:44.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I8eb7eb4f7a9ceed21401318e3395fe4450db13b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/41ac6d0509f292236f7253b2c8ba33440fb14a9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/41ac6d0509f292236f7253b2c8ba33440fb14a9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
