)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~654540","triplet_id":"openstack%2Fopenstack-ansible-os_zun~stable%2Fstein~Iab3b8260d7e3cd42842fd7bf435875157b273e18","project":"openstack/openstack-ansible-os_zun","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Iab3b8260d7e3cd42842fd7bf435875157b273e18","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 15:40:29.000000000","updated":"2019-04-22 18:34:11.000000000","submitted":"2019-04-22 18:34:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654540-1555958051838-795c03e7","meta_rev_id":"f22c2919807b255add8163da549886e563166337","_number":654540,"virtual_id_number":654540,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-22 18:34:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-22 16:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-22 16:18:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 16:18:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-22 18:34:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee0131f02f228cdc1b65849f9d8081fb5fdc2dd9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 15:40:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82c1940c3fca3ee41efc5f36a66f1dc94d0a6a0e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:18:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"085c9d9053c9c1e7b64a21fcccc916f16b2b3edd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:03:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/40/654540/1/check/openstack-ansible-linters/2a615a8/ : SUCCESS in 6m 06s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/40/654540/1/check/openstack-ansible-functional-debian-stable/51e2260/ : SUCCESS in 55m 58s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/40/654540/1/check/openstack-ansible-functional-centos-7/97d54b0/ : SUCCESS in 54m 17s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/40/654540/1/check/openstack-ansible-functional-ubuntu-bionic/2cdd742/ : SUCCESS in 42m 57s\n- openstack-tox-docs http://logs.openstack.org/40/654540/1/check/openstack-tox-docs/84ac87e/html/ : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3e925ca21647498447f77b8d31fc47f51dfb3161","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:03:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f705aa7913ac23a2d64fe2eb0b1a1fa404498b27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:34:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/40/654540/1/gate/openstack-ansible-linters/0553c66/ : SUCCESS in 4m 47s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/40/654540/1/gate/openstack-ansible-functional-debian-stable/02532ba/ : SUCCESS in 56m 59s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/40/654540/1/gate/openstack-ansible-functional-centos-7/8789349/ : SUCCESS in 58m 12s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/40/654540/1/gate/openstack-ansible-functional-ubuntu-bionic/27392d7/ : SUCCESS in 45m 44s\n- openstack-tox-docs http://logs.openstack.org/40/654540/1/gate/openstack-tox-docs/696bf79/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e3efb934d74d819073bbfc7465e5aaa482c5f39a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:34:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34e468ba320dc4eeaee6c63f186a6f2dd93a573a","revisions":{"34e468ba320dc4eeaee6c63f186a6f2dd93a573a":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:40:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/654540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/40/654540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/40/654540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/40/654540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/40/654540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/40/654540/1"}}},"commit":{"parents":[{"commit":"d39c44e3d6644a0a8a66c130826e63ca63787ffc","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/d39c44e3d6644a0a8a66c130826e63ca63787ffc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:40:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:40:27.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iab3b8260d7e3cd42842fd7bf435875157b273e18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/34e468ba320dc4eeaee6c63f186a6f2dd93a573a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/34e468ba320dc4eeaee6c63f186a6f2dd93a573a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
