)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~657270","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~Ib975a6d7e81bad87bc7025b0e682081ceeff8da1","project":"openstack/openstack-ansible-os_zun","branch":"master","hashtags":[],"change_id":"Ib975a6d7e81bad87bc7025b0e682081ceeff8da1","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-05-05 13:21:22.000000000","updated":"2019-06-27 22:54:03.000000000","submitted":"2019-06-27 22:54:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657270-1561676043595-be057232","meta_rev_id":"43b61275675fe5fef8de736586dd02e033505626","_number":657270,"virtual_id_number":657270,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-27 22:54:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-20 13:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-20 16:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-20 16:02:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 13:14:19.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-20 16:02:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-27 22:54:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72237a01bb8db76f0ba7a07db1d6cec2adc3a57b","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-05 13:21:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64bc3822027758603aa5a45fe947272521ec4d67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 14:27:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/70/657270/1/check/openstack-ansible-linters/487b819/ : SUCCESS in 4m 24s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/70/657270/1/check/openstack-ansible-functional-debian-stable/95e3561/ : SUCCESS in 48m 03s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/70/657270/1/check/openstack-ansible-functional-centos-7/676180e/ : SUCCESS in 52m 54s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/70/657270/1/check/openstack-ansible-functional-ubuntu-bionic/968bd08/ : SUCCESS in 44m 29s\n- openstack-tox-docs http://logs.openstack.org/70/657270/1/check/openstack-tox-docs/63bc246/html/ : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6e809adf49ce1ce6ae933c9cd5f8ae9a048c273f","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-06 10:11:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"063be125d03f50bd3a8b67bfda8ebe86772ae588","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 11:11:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/70/657270/2/check/openstack-ansible-linters/b2d1bcd/ : SUCCESS in 5m 24s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/70/657270/2/check/openstack-ansible-functional-debian-stable/424363f/ : SUCCESS in 53m 35s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/70/657270/2/check/openstack-ansible-functional-centos-7/81fb565/ : SUCCESS in 54m 13s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/70/657270/2/check/openstack-ansible-functional-ubuntu-bionic/d939aee/ : SUCCESS in 44m 14s\n- openstack-tox-docs http://logs.openstack.org/70/657270/2/check/openstack-tox-docs/736e872/html/ : SUCCESS in 2m 36s\n- build-openstack-releasenotes http://logs.openstack.org/70/657270/2/check/build-openstack-releasenotes/e6a854b/html/ : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":2},{"id":"04b89664cb7e05c98e232f9c5ed1acaf84db2a99","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-06 11:31:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ec5666075ce90e854c1136788bb57663e422410f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:33:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/70/657270/3/check/openstack-ansible-linters/730bf72/ : SUCCESS in 5m 27s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/70/657270/3/check/openstack-ansible-functional-debian-stable/a613166/ : SUCCESS in 57m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/70/657270/3/check/openstack-ansible-functional-centos-7/c331379/ : SUCCESS in 51m 50s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/70/657270/3/check/openstack-ansible-functional-ubuntu-bionic/9fd8307/ : SUCCESS in 45m 09s\n- openstack-tox-docs http://logs.openstack.org/70/657270/3/check/openstack-tox-docs/9b2075c/html/ : SUCCESS in 2m 14s\n- build-openstack-releasenotes http://logs.openstack.org/70/657270/3/check/build-openstack-releasenotes/a6d824b/html/ : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":3},{"id":"68c74a2e7a5a5777e49ec689d2e8784c6d86f80e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-20 13:14:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5a305130c4046a44c14d50edbc139b66ace7db84","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 16:02:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5ceaf44b009298f2ab4c7fcce9b755da692976be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 16:02:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e592e6fa3a3e7179425ef5f81565724bec06d96f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 17:27:24.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/70/657270/3/gate/openstack-ansible-linters/ac96451/ : SUCCESS in 6m 01s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/70/657270/3/gate/openstack-ansible-functional-debian-stable/58d0caf/ : SUCCESS in 57m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/70/657270/3/gate/openstack-ansible-functional-centos-7/2585328/ : FAILURE in 40m 41s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/70/657270/3/gate/openstack-ansible-functional-ubuntu-bionic/dc26295/ : SUCCESS in 1h 16m 48s\n- openstack-tox-docs http://logs.openstack.org/70/657270/3/gate/openstack-tox-docs/f79e15b/html/ : SUCCESS in 3m 59s\n- build-openstack-releasenotes http://logs.openstack.org/70/657270/3/gate/build-openstack-releasenotes/d49dc7b/html/ : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"2b26b583b2205e9d9efd4505b4e4bb6f98fc7da4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-21 10:53:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cc20013c406f2ad2168dc2ea800555924cdd8646","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 12:20:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/70/657270/3/check/openstack-ansible-linters/15a4b3a/ : SUCCESS in 5m 02s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/70/657270/3/check/openstack-ansible-functional-debian-stable/02e327e/ : SUCCESS in 56m 54s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/70/657270/3/check/openstack-ansible-functional-centos-7/1e05c37/ : FAILURE in 41m 35s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/70/657270/3/check/openstack-ansible-functional-ubuntu-bionic/0f5a9db/ : SUCCESS in 1h 01m 47s\n- openstack-tox-docs http://logs.openstack.org/70/657270/3/check/openstack-tox-docs/d77122d/html/ : SUCCESS in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/70/657270/3/check/build-openstack-releasenotes/c54a94f/html/ : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"934fceb786f6be8191e31bfecadb9e07ea2064e1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-27 20:31:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d6a9fc0ea2ca47eb785e6290af9408982dd2cc88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 21:34:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/70/657270/3/check/openstack-ansible-linters/b6c61af/ : SUCCESS in 4m 47s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/70/657270/3/check/openstack-ansible-functional-debian-stable/b0440b7/ : SUCCESS in 49m 03s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/70/657270/3/check/openstack-ansible-functional-centos-7/c685701/ : SUCCESS in 56m 11s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/70/657270/3/check/openstack-ansible-functional-ubuntu-bionic/8e38644/ : SUCCESS in 49m 33s\n- openstack-tox-docs http://logs.openstack.org/70/657270/3/check/openstack-tox-docs/f0342e2/html/ : SUCCESS in 2m 32s\n- build-openstack-releasenotes http://logs.openstack.org/70/657270/3/check/build-openstack-releasenotes/b2685ff/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":3},{"id":"70589f7e2535a8ebc39ecc50837786ab36e3994b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 21:36:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a87233710ee587ff30f4ab0d361e379bd2e7015b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 22:54:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/70/657270/3/gate/openstack-ansible-linters/4cb1eab/ : SUCCESS in 6m 59s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/70/657270/3/gate/openstack-ansible-functional-debian-stable/8ab31b9/ : SUCCESS in 57m 31s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/70/657270/3/gate/openstack-ansible-functional-centos-7/5822241/ : SUCCESS in 1h 11m 26s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/70/657270/3/gate/openstack-ansible-functional-ubuntu-bionic/54fb6bb/ : SUCCESS in 53m 04s\n- openstack-tox-docs http://logs.openstack.org/70/657270/3/gate/openstack-tox-docs/756bdc2/html/ : SUCCESS in 2m 53s\n- build-openstack-releasenotes http://logs.openstack.org/70/657270/3/gate/build-openstack-releasenotes/0084a0f/html/ : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"43b61275675fe5fef8de736586dd02e033505626","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 22:54:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5123fcda0f698e03a432ab46b7cc71475f6465fe","revisions":{"abfb4f369297274a3f7770b6590a7f227b2646be":{"kind":"REWORK","_number":1,"created":"2019-05-05 13:21:22.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/70/657270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/70/657270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/1"}}},"commit":{"parents":[{"commit":"41ac6d0509f292236f7253b2c8ba33440fb14a9e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/41ac6d0509f292236f7253b2c8ba33440fb14a9e"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:21:05.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:21:20.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ib975a6d7e81bad87bc7025b0e682081ceeff8da1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/abfb4f369297274a3f7770b6590a7f227b2646be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/abfb4f369297274a3f7770b6590a7f227b2646be"}]},"branch":"refs/heads/master"},"85272142ff626bce4681629b641e7af6fceb0be9":{"kind":"REWORK","_number":2,"created":"2019-05-06 10:11:31.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/70/657270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/70/657270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/2"}}},"commit":{"parents":[{"commit":"41ac6d0509f292236f7253b2c8ba33440fb14a9e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/41ac6d0509f292236f7253b2c8ba33440fb14a9e"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:21:05.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 10:11:16.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ib975a6d7e81bad87bc7025b0e682081ceeff8da1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/85272142ff626bce4681629b641e7af6fceb0be9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/85272142ff626bce4681629b641e7af6fceb0be9"}]},"branch":"refs/heads/master"},"5123fcda0f698e03a432ab46b7cc71475f6465fe":{"kind":"NO_CHANGE","_number":3,"created":"2019-05-06 11:31:03.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/70/657270/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/70/657270/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/70/657270/3"}}},"commit":{"parents":[{"commit":"41ac6d0509f292236f7253b2c8ba33440fb14a9e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/41ac6d0509f292236f7253b2c8ba33440fb14a9e"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:21:05.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 11:30:47.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ib975a6d7e81bad87bc7025b0e682081ceeff8da1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/5123fcda0f698e03a432ab46b7cc71475f6465fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/5123fcda0f698e03a432ab46b7cc71475f6465fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
