)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~658043","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~I2c70a99d005eb4d7879de0105c924393f01b6288","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I2c70a99d005eb4d7879de0105c924393f01b6288","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:35:48.000000000","updated":"2019-05-09 14:56:08.000000000","submitted":"2019-05-09 14:56:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658043-1557413768899-21cd92bb","meta_rev_id":"6f198a3e58f8b06f32958665162a155a4992e11c","_number":658043,"virtual_id_number":658043,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-09 14:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-09 12:55:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-09 12:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:55:55.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-09 14:56:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14e77a4003222583ce81070bb0e6c15f272df421","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:35:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97bc3f9d150177c9b423a8a19b7c67a50e68c76b","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-09 12:55:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"908da1c3475fc4c3107b09135759274892f0d01d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:40:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/43/658043/1/check/openstack-ansible-linters/fa99aa7/ : SUCCESS in 5m 18s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/43/658043/1/check/openstack-ansible-functional-debian-stable/5f857c4/ : SUCCESS in 59m 05s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/43/658043/1/check/openstack-ansible-functional-centos-7/fc62e96/ : SUCCESS in 1h 38m 06s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/43/658043/1/check/openstack-ansible-functional-ubuntu-bionic/682febf/ : SUCCESS in 57m 00s\n- openstack-tox-docs http://logs.openstack.org/43/658043/1/check/openstack-tox-docs/43f9f85/html/ : SUCCESS in 3m 43s\n- build-openstack-releasenotes http://logs.openstack.org/43/658043/1/check/build-openstack-releasenotes/66d3356/html/ : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d4080498212e8aabb0fd4a6f34af7e7f69d09bf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:40:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa82e31108db53a76f6fadafe2d4a056178b70d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 14:56:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e4a90e928794eb8c0fa80315b8c4010cc88cf844","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 14:56:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/43/658043/1/gate/openstack-ansible-linters/6b42fff/ : SUCCESS in 4m 41s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/43/658043/1/gate/openstack-ansible-functional-debian-stable/060372c/ : SUCCESS in 59m 31s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/43/658043/1/gate/openstack-ansible-functional-centos-7/146b1e8/ : SUCCESS in 52m 49s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/43/658043/1/gate/openstack-ansible-functional-ubuntu-bionic/88837c8/ : SUCCESS in 48m 41s\n- openstack-tox-docs http://logs.openstack.org/43/658043/1/gate/openstack-tox-docs/5229036/html/ : SUCCESS in 2m 32s\n- build-openstack-releasenotes http://logs.openstack.org/43/658043/1/gate/build-openstack-releasenotes/14be5ab/html/ : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7a91ed14a77cb07918d86605b4fc96a7e39d99b","revisions":{"d7a91ed14a77cb07918d86605b4fc96a7e39d99b":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:35:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/658043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/43/658043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/658043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/658043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/658043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/658043/1"}}},"commit":{"parents":[{"commit":"41ac6d0509f292236f7253b2c8ba33440fb14a9e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/41ac6d0509f292236f7253b2c8ba33440fb14a9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:35:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:35:47.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2c70a99d005eb4d7879de0105c924393f01b6288\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/d7a91ed14a77cb07918d86605b4fc96a7e39d99b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/d7a91ed14a77cb07918d86605b4fc96a7e39d99b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
