)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~667074","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~I441feb4b74b72002af0284137e6025821024979e","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"osa/db-refactor","hashtags":[],"change_id":"I441feb4b74b72002af0284137e6025821024979e","subject":"db_setup: refactor database setup to a common file","status":"MERGED","created":"2019-06-24 12:27:22.000000000","updated":"2019-06-27 13:34:32.000000000","submitted":"2019-06-27 13:34:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667074-1561642472121-328d8c3f","meta_rev_id":"de892e7fdfa93cbfb39ab893f9b0147f8201b3a9","_number":667074,"virtual_id_number":667074,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-06-27 13:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-06-24 21:02:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-27 12:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-27 12:30:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 21:02:36.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-06-27 12:30:26.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-27 13:34:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"680b661b5c647296c8137f3c013b80ed43fc3a24","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-24 12:27:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcf8abe7d2ea1f92461d5b9454c5cf56286f0fba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 13:26:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/74/667074/1/check/openstack-ansible-linters/38a0ab1/ : SUCCESS in 5m 42s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/74/667074/1/check/openstack-ansible-functional-debian-stable/66334f6/ : SUCCESS in 54m 02s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/667074/1/check/openstack-ansible-functional-centos-7/216eb32/ : FAILURE in 39m 42s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/74/667074/1/check/openstack-ansible-functional-ubuntu-bionic/5ba416b/ : SUCCESS in 47m 35s\n- openstack-tox-docs http://logs.openstack.org/74/667074/1/check/openstack-tox-docs/3de926a/html/ : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3c8e3e3437fdbef3f0688c95de428f577139a72d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-24 18:17:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5c19c85c47fdb8381c3e12850016c549e367f3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 19:19:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/74/667074/2/check/openstack-ansible-linters/494f252/ : SUCCESS in 5m 02s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/74/667074/2/check/openstack-ansible-functional-debian-stable/35a6d5d/ : SUCCESS in 49m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/667074/2/check/openstack-ansible-functional-centos-7/b7cb21c/ : SUCCESS in 52m 44s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/74/667074/2/check/openstack-ansible-functional-ubuntu-bionic/51a830a/ : SUCCESS in 57m 13s\n- openstack-tox-docs http://logs.openstack.org/74/667074/2/check/openstack-tox-docs/27ccf3e/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"409fc51a13e9b0db6fba360e74650b385cde93a5","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-06-24 21:02:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e7c100cfc7e2c87efbf3c3a454f2a7413d860bf3","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-27 12:30:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"34e4e0b236f18f64b71305bf38aa670627517616","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:30:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a387d68a4a35721dcca321e38444d91df5e2279","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:34:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/74/667074/2/gate/openstack-ansible-linters/b79aea3/ : SUCCESS in 5m 28s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/74/667074/2/gate/openstack-ansible-functional-debian-stable/741da64/ : SUCCESS in 57m 20s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/667074/2/gate/openstack-ansible-functional-centos-7/72a7fa7/ : SUCCESS in 57m 19s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/74/667074/2/gate/openstack-ansible-functional-ubuntu-bionic/b6fd9c9/ : SUCCESS in 44m 47s\n- openstack-tox-docs http://logs.openstack.org/74/667074/2/gate/openstack-tox-docs/f757125/html/ : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"c689652f90b88ea073768fed8ff51a511539b2a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:34:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0475d4218ea9cc5961811b887f5bd77856727331","revisions":{"2aca4dcf56d8d15abed5bebcb4b120d2bd85a04c":{"kind":"REWORK","_number":1,"created":"2019-06-24 12:27:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/667074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/74/667074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/74/667074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/74/667074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/74/667074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/74/667074/1"}}},"commit":{"parents":[{"commit":"4d906711ec708698ee3e260b609a60c815e880fc","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/4d906711ec708698ee3e260b609a60c815e880fc"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-06-24 12:27:10.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-06-24 12:27:10.000000000","tz":180},"subject":"db_setup: refactor database setup to a common file","message":"db_setup: refactor database setup to a common file\n\nThis patch refactors the database creation to db_setup.yml which\nwill eventually be managed by openstack-ansible-tests.\n\nChange-Id: I441feb4b74b72002af0284137e6025821024979e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/2aca4dcf56d8d15abed5bebcb4b120d2bd85a04c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/2aca4dcf56d8d15abed5bebcb4b120d2bd85a04c"}]},"branch":"refs/heads/master"},"0475d4218ea9cc5961811b887f5bd77856727331":{"kind":"REWORK","_number":2,"created":"2019-06-24 18:17:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/667074/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/74/667074/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/74/667074/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/74/667074/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/74/667074/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/74/667074/2"}}},"commit":{"parents":[{"commit":"4d906711ec708698ee3e260b609a60c815e880fc","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/4d906711ec708698ee3e260b609a60c815e880fc"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-06-24 12:27:10.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-06-24 18:16:21.000000000","tz":180},"subject":"db_setup: refactor database setup to a common file","message":"db_setup: refactor database setup to a common file\n\nThis patch refactors the database creation to db_setup.yml which\nwill eventually be managed by openstack-ansible-tests.\n\nAdditionally enabling extras repo for centos, since it\u0027s not\nenabled by default but required by docker\n\nChange-Id: I441feb4b74b72002af0284137e6025821024979e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/0475d4218ea9cc5961811b887f5bd77856727331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/0475d4218ea9cc5961811b887f5bd77856727331"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
