)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~689528","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~I2efc775d7e38019eb39ca85b6135b8506fc76c52","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"zun_tests","hashtags":[],"change_id":"I2efc775d7e38019eb39ca85b6135b8506fc76c52","subject":"Add uwsgi role into local a-r-r and test placement","status":"MERGED","created":"2019-10-18 15:42:49.000000000","updated":"2019-11-08 08:00:34.000000000","submitted":"2019-11-07 23:46:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689528-1573170372096-c91a571c","meta_rev_id":"520f0a0f7f9c7a63a428654b1cfa64140eabbbb3","_number":689528,"virtual_id_number":689528,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-07 23:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-11-07 18:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-07 22:33:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-07 22:33:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-07 18:08:19.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-11-07 22:33:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-07 23:46:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40ab324b1d6bee5df9f03ca92c54f02365d54aed","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-18 15:42:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f959edb6b1d6304ca3c7d1dfdd950441e6862ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 16:59:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/15ba5f8214e2450d8a7ad2ac53605975 : SUCCESS in 6m 31s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/3619ec398c8c43a0b52f51bba69784d0 : FAILURE in 1h 09m 43s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/57b58f0c2a2c4b25a44c93ee3f50d2f4 : FAILURE in 1h 09m 54s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d109e3b5229400c92885049428b9bfc : FAILURE in 51m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5905d8b83ff4f4c9a74da1f50850e3d : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"769ccf6762661bece52f5a78d35e4af30c067940","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-18 17:27:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"79309c7e208065464403d677fed0268f4108dff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 17:42:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3fac80a0f7cb4073aa3952d91552d2c5 : FAILURE in 4m 32s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/973ec1f1d14b46e78b8765f22769d157 : FAILURE in 4m 52s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c03b0913c083419eb7282772d6cad670 : FAILURE in 11m 23s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/451c4937634849ada4846443252b7af4 : FAILURE in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13db92893904933ac965c8c8ec0d843 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"a3af23064fbe8f5e37077df56ae3929c24e63ab8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-18 18:24:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"21a6a03f7e98c552c6641fc7dbebc08e3d6413a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 20:00:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8fb977baf5584549b5da4901bdbefa85 : SUCCESS in 6m 35s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/0503274e85de44efa51b02650a6e5b00 : FAILURE in 23m 48s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e4784187026e4b78aba224c9f92f0eb5 : FAILURE in 28m 20s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e5d91e4cf9e40aeb4440d041097aec3 : FAILURE in 26m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ee5e63a73d484a9cd984227976e98b : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"42b6b3e4c4bec40dd1b85e6d3206ba514bdbe9a0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-31 16:42:15.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"941ff3561186cd92137ee36e00664f27698778c9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-07 14:57:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"96519766347951531262f1d9e10af1874b4e1bb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 16:23:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c65298e7897d4d6481c5e3779fb14f65 : SUCCESS in 5m 40s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/9f27195b816a4f0ea1ed7e9c7e1fe322 : SUCCESS in 1h 05m 30s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/6e1f9ddbf5c14110aec386f6bc8c5390 : SUCCESS in 1h 20m 06s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e130c5b7efa640ceaf08feb98a03dfea : SUCCESS in 51m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f554c9f4e44b75ae3d48f36519907c : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":4},{"id":"9ac7cb033dde5ffbafe470568823df5d87ceba54","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-11-07 18:08:19.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0b3283090e33659f6032c16c7757285d6049e9a2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-07 22:33:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"461324132b30d92eabee67c4315118ac82288698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 22:33:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c23ab95f488adddc0c5861634b910fcaef61fc23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 23:46:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5af18abfa4bd49dca2972155d7cc2112 : SUCCESS in 7m 18s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/aa3822bff39a49b08d4d67b39ac3aa8f : SUCCESS in 1h 12m 02s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/edc2f1c68e1f416c87a5596008b3e93f : SUCCESS in 57m 26s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb64f7f8463f48baa297ab4c78c32335 : SUCCESS in 51m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd69e7b818614ed68000094211001cdf : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":4},{"id":"f8f38d2804dc9b2dc36c5431a12926748ad6f3d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 23:46:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ed1f76108ac5a5f76a81bd76b163b702f832edf9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 23:47:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2371d4047b8e4b909e552189073665d1 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4},{"id":"d06b1541253728aa336666f3066d7941db28b59a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-08 08:00:34.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit e91069b5dbf795863c2288760e606adb23c6e69d","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7cb2d444c06cb9306d4e1d045603635cd8d8870d","revisions":{"3c896f3f231416c8a67352bad08fefc1e3edf404":{"kind":"REWORK","_number":1,"created":"2019-10-18 15:42:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/689528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/28/689528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/1"}}},"commit":{"parents":[{"commit":"0f95ff232f4b973cc71b40f083a0eb5cd4d05918","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/0f95ff232f4b973cc71b40f083a0eb5cd4d05918"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-18 15:42:39.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-18 15:42:39.000000000","tz":180},"subject":"Add uwsgi role into local a-r-r","message":"Add uwsgi role into local a-r-r\n\nChange-Id: I2efc775d7e38019eb39ca85b6135b8506fc76c52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/3c896f3f231416c8a67352bad08fefc1e3edf404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/3c896f3f231416c8a67352bad08fefc1e3edf404"}]},"branch":"refs/heads/master"},"f89e19e94882fdc89931566ac564069362659ffa":{"kind":"REWORK","_number":2,"created":"2019-10-18 17:27:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/689528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/28/689528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/2"}}},"commit":{"parents":[{"commit":"0f95ff232f4b973cc71b40f083a0eb5cd4d05918","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/0f95ff232f4b973cc71b40f083a0eb5cd4d05918"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-18 17:27:12.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-18 17:27:12.000000000","tz":180},"subject":"Add uwsgi role into local a-r-r and test placement","message":"Add uwsgi role into local a-r-r and test placement\n\nChange-Id: I2efc775d7e38019eb39ca85b6135b8506fc76c52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/f89e19e94882fdc89931566ac564069362659ffa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/f89e19e94882fdc89931566ac564069362659ffa"}]},"branch":"refs/heads/master"},"8923041a20848f36a4101cafb8f3ca931e9dd84d":{"kind":"REWORK","_number":3,"created":"2019-10-18 18:24:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/689528/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/28/689528/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/3"}}},"commit":{"parents":[{"commit":"0f95ff232f4b973cc71b40f083a0eb5cd4d05918","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/0f95ff232f4b973cc71b40f083a0eb5cd4d05918"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-18 17:27:12.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-18 18:24:46.000000000","tz":180},"subject":"Add uwsgi role into local a-r-r and test placement","message":"Add uwsgi role into local a-r-r and test placement\n\nChange-Id: I2efc775d7e38019eb39ca85b6135b8506fc76c52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/8923041a20848f36a4101cafb8f3ca931e9dd84d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/8923041a20848f36a4101cafb8f3ca931e9dd84d"}]},"branch":"refs/heads/master"},"7cb2d444c06cb9306d4e1d045603635cd8d8870d":{"kind":"REWORK","_number":4,"created":"2019-11-07 14:57:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/689528/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/28/689528/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/28/689528/4"}}},"commit":{"parents":[{"commit":"0f95ff232f4b973cc71b40f083a0eb5cd4d05918","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/0f95ff232f4b973cc71b40f083a0eb5cd4d05918"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-18 17:27:12.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-07 14:57:10.000000000","tz":120},"subject":"Add uwsgi role into local a-r-r and test placement","message":"Add uwsgi role into local a-r-r and test placement\n\nChange-Id: I2efc775d7e38019eb39ca85b6135b8506fc76c52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/7cb2d444c06cb9306d4e1d045603635cd8d8870d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/7cb2d444c06cb9306d4e1d045603635cd8d8870d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
