)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~693543","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~I1d2316fbe9ae0c74b9d516d3a143c7a58ff59365","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"use-integrated-repo-for-all-tests","hashtags":[],"change_id":"I1d2316fbe9ae0c74b9d516d3a143c7a58ff59365","subject":"Drop commented out values from configs","status":"MERGED","created":"2019-11-08 14:37:22.000000000","updated":"2020-04-02 18:37:15.000000000","submitted":"2020-04-02 18:23:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693543-1585851819730-6894be15","meta_rev_id":"1a5f7ed3ad1616adfb7ea19defd20bba0f57581d","_number":693543,"virtual_id_number":693543,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"date":"2020-03-29 01:21:55.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 18:23:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-04-02 15:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-29 10:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2020-04-02 15:25:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 14:05:49.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-03-29 01:21:55.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-03-29 10:02:49.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-02 15:25:52.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-04-02 18:23:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfd3d0dbf26e9b4e418aa31e47a264e3b03de10a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-08 14:37:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d0520a3be099130f2c318c08e74369ab49133b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 15:45:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9f0fe3779dba40ea91d968d1654d7c6e : SUCCESS in 6m 27s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/b3cf22a9c869449a9918c95b365e0567 : FAILURE in 40m 17s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/51a1f3fb625948f7be4444356825d1da : FAILURE in 49m 44s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bad5f1f414cf4817b28a00898f334ff1 : FAILURE in 49m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3330030622b74ba083096252f92c26c6 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cb359917c6f6f07b7d02b4016ecd2cdc083b2989","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-11 13:34:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4a748add1b90522b0afa5919df8013f19a0b0568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 15:05:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/37cc6498dc3d4b1998bfb2d76bef4081 : SUCCESS in 4m 15s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/59ce26c4210c415c900c3b01b1a29575 : SUCCESS in 1h 06m 36s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2c1f20da384e461a9a103878cf5ebd44 : SUCCESS in 57m 46s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8657ad576aca481bad0db184ee2e4942 : SUCCESS in 51m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f259e756adb441faef2fbdd821d7308 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":2},{"id":"5c83d20feeec63bc737325c2a92779588f673c3b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-11 17:29:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9c2c470893cc703a9ea17b849bf3fb832ca6f84a","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-11-12 14:05:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5aedb27aa646223a3b1e7c7c394c20e362763675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 14:06:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2fd3ef60e456869733932c6f54d4a77c5b9c9c68","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-12 14:10:48.000000000","message":"Patch Set 2: Workflow-1\n\nJust found out that when host_shared_with_nova is true, zun get wrong hostname. So either host should be set in conf, or zun edited","accounts_in_message":[],"_revision_number":2},{"id":"014990e770b9aebca54fd1d8abcac144e48734bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 15:14:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dd005607da1a438185ef2629e821bd80 : SUCCESS in 5m 29s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/2c3c0d6e26394109ae48a38d4652c39f : FAILURE in 15m 04s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4fb2551408534fb895ede42e4c314a5a : SUCCESS in 1h 03m 37s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e0e7871c49a432f91053f6dc5bf8d95 : SUCCESS in 1h 02m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ebdf32145d4edd937897f9885dd0ab : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ec8530072b01dd8d646a5cff81341a585a11406c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-12 18:35:07.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0373880c8b3ef0ea99deab7b7f8c29c2c5ad1630","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-12 18:35:12.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"b003acbb744f125b2b639ec61bc783d54502b889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 20:15:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e2b82cea1b544d8d959e88281bbb40e3 : SUCCESS in 5m 33s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/cdb258732eba414cae08a87918f6ca99 : SUCCESS in 1h 14m 01s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/707da7c28d7348f8b3aba4416ceea1b5 : SUCCESS in 59m 48s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c690c476ca4540ebb541c3425fd72ff2 : SUCCESS in 55m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c34411cb1c804951bd1c52524485ae17 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":3},{"id":"da7eaae8134a5f1c08aef6dc6c4b6f2e9439bbb7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-26 13:01:52.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"76566eb7522e712e862f395f4d034d3ac384540e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-26 13:01:59.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"5f3e995bbf96c55b62294bf945d68d7176e286ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 14:37:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/16dd0a5144164bdaa3255e4b40d52132 : SUCCESS in 9m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/25bdf2d1442749098dbdb6ca9d171e14 : FAILURE in 1h 04m 46s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/64aebc47277d40fba4bfe0a4dd6780f0 : FAILURE in 1h 03m 58s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ed3ffa413c24ded8e29e088bff5472d : FAILURE in 1h 31m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/119f2fb0874b434d81bfe691db08b2c6 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":4},{"id":"a8856eda668a33ea4b81132bd75f9c1ca1aae810","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-26 19:11:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a0c6d451627bbb765e20f7b5d97b38548384e6c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 21:20:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5131aa1f4ca44faabed93a97fb554594 : SUCCESS in 10m 32s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9090afbb1b744986a83030738c04dfe2 : SUCCESS in 1h 11m 00s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/f1cad5f691944eeca1a085c62969a150 : SUCCESS in 1h 00m 49s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3605ae56a4644d1cbe1706ff16f26848 : SUCCESS in 57m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6567376198314440aa641aeddbfa29a8 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":5},{"id":"7214acbc8c56c372bd993298b140ff23b31582fd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-29 10:02:49.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bbd100862339190586e45f8bdc0c83ede365b756","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-04-02 15:25:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9cfb816ca8fbaf40c019edabc11a5a721cc440ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:20:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a874b407d961e9746485fc138a135fc3836b3086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:23:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eb07c30cd9734335b969d2a8f8b49f12 : SUCCESS in 12m 23s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/01de593148fc4b25922fe6c48a0be049 : SUCCESS in 1h 04m 44s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e75b07f62d4643209240a758ab9dc275 : SUCCESS in 1h 02m 00s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/753a02ef011641ed9aa89604e5297bb6 : SUCCESS in 1h 00m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25fc9c1de344612bd5bbe6f368c4e02 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":5},{"id":"5ff5bc7a3ffe27315eebc214105c2026cd75ee2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:23:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"eda2d474e24ae749c1d9a0dbd4d2ea1f613054ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:37:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d8702802764f49b307a89567f7a0c1 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a240a29b80b38e2ea138f00df9d30139f4e18b6d","revisions":{"7b62585b2a2eb06db1dd5a1042b7d12eb9e3c3f7":{"kind":"REWORK","_number":1,"created":"2019-11-08 14:37:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/693543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/43/693543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/1"}}},"commit":{"parents":[{"commit":"bc39aac81b58bf60a44ce051d1820fce44771a81","subject":"Start using uWSGI role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/bc39aac81b58bf60a44ce051d1820fce44771a81"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-08 11:58:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-08 14:36:56.000000000","tz":120},"subject":"Drop commented out values from configs","message":"Drop commented out values from configs\n\nTo make configs more readable it\u0027s worth dropping out commented out\nlines. Also some issues in configs has been fixed.\n\nChange-Id: I1d2316fbe9ae0c74b9d516d3a143c7a58ff59365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/7b62585b2a2eb06db1dd5a1042b7d12eb9e3c3f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/7b62585b2a2eb06db1dd5a1042b7d12eb9e3c3f7"}]},"branch":"refs/heads/master"},"3cf018f9412ac64bc1c5e5f4832a918eb9d099dc":{"kind":"REWORK","_number":2,"created":"2019-11-11 13:34:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/693543/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/43/693543/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/2"}}},"commit":{"parents":[{"commit":"bc39aac81b58bf60a44ce051d1820fce44771a81","subject":"Start using uWSGI role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/bc39aac81b58bf60a44ce051d1820fce44771a81"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-08 11:58:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-11-11 13:34:44.000000000","tz":0},"subject":"Drop commented out values from configs","message":"Drop commented out values from configs\n\nTo make configs more readable it\u0027s worth dropping out commented out\nlines. Also some issues in configs has been fixed.\n\nChange-Id: I1d2316fbe9ae0c74b9d516d3a143c7a58ff59365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/3cf018f9412ac64bc1c5e5f4832a918eb9d099dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/3cf018f9412ac64bc1c5e5f4832a918eb9d099dc"}]},"branch":"refs/heads/master"},"23d00c5d29e72a047dcf1a54796ed96abb226ce1":{"kind":"REWORK","_number":3,"created":"2019-11-12 18:35:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/693543/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/43/693543/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/3"}}},"commit":{"parents":[{"commit":"bc39aac81b58bf60a44ce051d1820fce44771a81","subject":"Start using uWSGI role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/bc39aac81b58bf60a44ce051d1820fce44771a81"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-08 11:58:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-11-12 18:35:03.000000000","tz":0},"subject":"Drop commented out values from configs","message":"Drop commented out values from configs\n\nTo make configs more readable it\u0027s worth dropping out commented out\nlines. Also some issues in configs has been fixed.\n\nChange-Id: I1d2316fbe9ae0c74b9d516d3a143c7a58ff59365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/23d00c5d29e72a047dcf1a54796ed96abb226ce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/23d00c5d29e72a047dcf1a54796ed96abb226ce1"}]},"branch":"refs/heads/master"},"2d1cb347dbaf6ae57df03959ef9255843b1f4c42":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-26 13:01:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/693543/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/43/693543/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/4"}}},"commit":{"parents":[{"commit":"399db8a94e5dd62c1ba3f7f08e183e4405666517","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/399db8a94e5dd62c1ba3f7f08e183e4405666517"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-08 11:58:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-03-26 13:01:52.000000000","tz":0},"subject":"Drop commented out values from configs","message":"Drop commented out values from configs\n\nTo make configs more readable it\u0027s worth dropping out commented out\nlines. Also some issues in configs has been fixed.\n\nChange-Id: I1d2316fbe9ae0c74b9d516d3a143c7a58ff59365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/2d1cb347dbaf6ae57df03959ef9255843b1f4c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/2d1cb347dbaf6ae57df03959ef9255843b1f4c42"}]},"branch":"refs/heads/master"},"a240a29b80b38e2ea138f00df9d30139f4e18b6d":{"kind":"REWORK","_number":5,"created":"2020-03-26 19:11:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/693543/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/43/693543/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/43/693543/5"}}},"commit":{"parents":[{"commit":"7ee1939c1cd6e7850817988d6497438c3d0dde10","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/7ee1939c1cd6e7850817988d6497438c3d0dde10"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-08 11:58:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-03-26 19:11:29.000000000","tz":120},"subject":"Drop commented out values from configs","message":"Drop commented out values from configs\n\nTo make configs more readable it\u0027s worth dropping out commented out\nlines. Also some issues in configs has been fixed.\n\nChange-Id: I1d2316fbe9ae0c74b9d516d3a143c7a58ff59365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/a240a29b80b38e2ea138f00df9d30139f4e18b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/a240a29b80b38e2ea138f00df9d30139f4e18b6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
