)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~709756","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~Ib1569f749152832e00afdfa49a1f7ddc6176a637","project":"openstack/openstack-ansible-os_zun","branch":"master","hashtags":[],"change_id":"Ib1569f749152832e00afdfa49a1f7ddc6176a637","subject":"Remove global python interpreter setting.","status":"ABANDONED","created":"2020-02-25 15:22:21.000000000","updated":"2022-01-04 10:38:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c5c8e5f38b807572faec5e825a1d4b8cdafc7692","_number":709756,"virtual_id_number":709756,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-25 16:14:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 16:14:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d659c48634019bfab09c17f58fad364e177a02f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 15:22:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6deff42432b8c5ab371bc6f2d6d7e19a8b7c270c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:14:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/52f738e1bf2646bdae3cddebc3d9244c : SUCCESS in 7m 48s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/41801ba90af841eebed66451673a356c : FAILURE in 34m 55s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/6bbc2b7ab0784e31bc42a3ce459fc655 : FAILURE in 45m 07s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2a9d90bb0584d799cf21427b8715654 : FAILURE in 36m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ebef73910bc4a63b5cdbd21e02b453e : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c5c8e5f38b807572faec5e825a1d4b8cdafc7692","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-04 10:38:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b663fcd1f64b244a39ed86742dffe8ed02dc6346","revisions":{"b663fcd1f64b244a39ed86742dffe8ed02dc6346":{"kind":"REWORK","_number":1,"created":"2020-02-25 15:22:21.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/56/709756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/56/709756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/56/709756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/56/709756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/56/709756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/56/709756/1"}}},"commit":{"parents":[{"commit":"399db8a94e5dd62c1ba3f7f08e183e4405666517","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/399db8a94e5dd62c1ba3f7f08e183e4405666517"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-25 15:21:29.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-25 15:22:18.000000000","tz":0},"subject":"Remove global python interpreter setting.","message":"Remove global python interpreter setting.\n\nWe now default to python3 so reduce the number of variables\n\nChange-Id: Ib1569f749152832e00afdfa49a1f7ddc6176a637\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/b663fcd1f64b244a39ed86742dffe8ed02dc6346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/b663fcd1f64b244a39ed86742dffe8ed02dc6346"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
