)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~710685","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~I238a0342269ce3e71050594425ed2c5876ce9b54","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I238a0342269ce3e71050594425ed2c5876ce9b54","subject":"[ussuri][goal] Updates for python 2.7 drop","status":"MERGED","created":"2020-03-02 01:23:44.000000000","updated":"2020-05-12 10:46:59.000000000","submitted":"2020-05-12 10:44:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710685-1589280276692-354c625b","meta_rev_id":"106d801b705dc2d63b46d98e6525678ad7baf654","_number":710685,"virtual_id_number":710685,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-11 11:53:08.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 10:44:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-05-04 09:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-01 17:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-05-04 09:42:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-01 17:01:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-04 09:42:39.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-05-11 11:53:08.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-12 10:44:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf6fbd7c7790c0dc85194ffcd3b7c3dc34a59336","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-02 01:23:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72b9f927a3cca6db4e39786833c7a2582b9e0f43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 02:20:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9fd23e7882714d4da109f8b7a17c8551 : SUCCESS in 7m 52s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9e98b1c9b3ad471082bc3812c63862dc : FAILURE in 38m 09s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/5eb319577bf14d6295e3f18b50c11295 : FAILURE in 5m 39s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f89452d7780a4715b9dedb43e11eb9a7 : SUCCESS in 55m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b74d5de27f4e84a0cb43b1993f1fa5 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2156b12536f24e7cb9de285ede0cae94 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4ae167336c0bbd61f5354cbf84d1ce2a93de9b98","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 18:48:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3c02ce63e8f9ce17be26adc05c8eb2c0dc72c599","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:58:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c60c950f863e4d1dae591125786df56a : SUCCESS in 8m 05s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c2c2b027a9fd45e38a9d6f18443b5057 : SUCCESS in 1h 07m 22s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c809135db5b34005b9a857b7315e705e : FAILURE in 25m 45s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08f6691c20a14fff935702cbf5dd6db7 : SUCCESS in 1h 04m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28805f5f36b0439cbeb0f2cb96eddea9 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d20ef8a52234473a56c7c8441850558 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"95bad7f2eaea12ffcbafe5a17c8c308602522f04","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-29 06:17:48.000000000","message":"Patch Set 1:\n\nrecheck vexxhost mirror error for centos","accounts_in_message":[],"_revision_number":1},{"id":"6711241ae253e799ce68b46417054878be0741ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 07:37:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/556fefd504d24217838a16a95ad1aca8 : SUCCESS in 9m 53s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/3e7c97b06af24c17b7e2dc9ab089e636 : SUCCESS in 1h 09m 54s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ab2cd94ed0a84eb48ba95c15bb9429b9 : SUCCESS in 1h 11m 40s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccee6ce443be4f8ea33a2c17102e9cd5 : SUCCESS in 58m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d2ae5821b85418ab881a28d902ab2d0 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96d8ccb24c094a648d27b0529dca66bd : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"dd30c02654741243c01d66ddfddbbbc22ef7e51a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-01 17:01:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f04cdfb60beadff33681ca8fe1bc3cf22b42d77e","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-05-04 09:42:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4daa2bcd4e1763b9fecf0ebb0beeb46588a8d9ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 09:42:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bff0cd6e6d33f3c2b82b9cbdcfcaa641b6d5dad7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 10:37:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b18910444d8d40b4ac1b1139d880fb65 : SUCCESS in 6m 54s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/6b39fc4e977f4714ac89fe726b5d73e9 : FAILURE in 44m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0d40bf3055184eaea95a8badd12b3597 : FAILURE in 49m 40s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7fbf22aa99247a8b395fd3bc783ce93 : FAILURE in 43m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463ee9ab2e884181ace90e9114e8845a : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a32d2f16e9564a699984581633b4ef70 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e26d7bfed87524d1679119ed6c8510dca0730a3c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 15:15:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"17e309aa2e9de03bd6654464116dc074299889f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:27:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8bc8bd7602534e9ba602533a0196c147 : SUCCESS in 7m 50s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b77828b96aaf443ea83c4019bc58aa58 : FAILURE in 48m 31s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a74af9df28cb4d96a804abb4ee3063a5 : FAILURE in 1h 10m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db6bdc17f50a45eca34371504386d8d9 : FAILURE in 1h 07m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d48c5307ba45078c105efd9c0dc82a : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c26bde1b3fb54c98afffb839fb854fa5 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"856f5ac965b931f36832d78d89826ca79d0cb5dd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-11 11:53:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88c379a0590eae8a1fa2b8aa20fb7f29ed7cfe55","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-11 12:02:51.000000000","message":"Patch Set 1:\n\nlooks like we need something like this https://github.com/logan2211/ansible-etcd/pull/18","accounts_in_message":[],"_revision_number":1},{"id":"4212c0c7586f7d475dd56590213914140930aba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 12:47:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/db8861548379453bbe31044f7a49098f : SUCCESS in 10m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1afcfefdb34f4c8c82936ebefd8a736c : FAILURE in 50m 14s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d730e9c4f087414a8564d6c3d4a04acd : FAILURE in 52m 01s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/531a92826911411fb4ea805ab8657145 : FAILURE in 49m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec7906cbe55469fb3bd0a1b7bef5027 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7e798f9b02e441fa19280ce7024f37e : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"cdc26f85e7b66e27cd31ef34fce19d4cd33c375b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-12 07:41:29.000000000","message":"Patch Set 1:\n\nrecheck merged etc ansible role PR https://github.com/logan2211/ansible-etcd/pull/18","accounts_in_message":[],"_revision_number":1},{"id":"9009641e69cc6425c329805394fe507e03eccad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:15:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0ec69b8330ad43718e626ed482c35cac : SUCCESS in 8m 49s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2e16c04633664cd7b423093d9d718985 : SUCCESS in 1h 10m 19s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/f7bb20c008da42b9883712f84a9102fb : SUCCESS in 1h 05m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2e9c28bb7434986ba66a8f02c512f96 : SUCCESS in 59m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a7bee37bee441ea9048c7d5569d239e : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/996ec8d5860e42ff84c4f9300f55a262 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3b749f4dd32252f3be2eecf5a156480eef6a6d63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:15:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6fbf4d257eb27a85a29d3bb6384960abf3096a4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 10:44:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7a093fc64f27a26d378a1ee878c3746776be9dd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 10:44:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0389a36a9f43434aacdf69f7dc629bda : SUCCESS in 6m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4811e2881bf848ca9b58a5074231ed68 : SUCCESS in 1h 06m 53s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8ffdfcfa64184be59f08e60a41785013 : SUCCESS in 1h 04m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28e78ef6f7e745b5bfd6fbbcc272e7a0 : SUCCESS in 1h 07m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/154868206cd7455497a9c1fe04effcdb : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73f439b546da4dadb2e47cdb4448a456 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"018fe413206ee4fdc0e4e86465ec9a9e9c4cb60c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 10:46:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6dfff036d334ea8be6543e8617871b0 : SUCCESS in 2m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84a4dba1b03443338b11f6285b4f9331 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2ee23a7b0173c99f66ea34163c37de5dd070a16","revisions":{"e2ee23a7b0173c99f66ea34163c37de5dd070a16":{"kind":"REWORK","_number":1,"created":"2020-03-02 01:23:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/85/710685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/85/710685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/710685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/710685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/710685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/710685/1"}}},"commit":{"parents":[{"commit":"399db8a94e5dd62c1ba3f7f08e183e4405666517","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/399db8a94e5dd62c1ba3f7f08e183e4405666517"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-02 01:14:48.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-02 01:23:42.000000000","tz":-360},"subject":"[ussuri][goal] Updates for python 2.7 drop","message":"[ussuri][goal] Updates for python 2.7 drop\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nOnly update required is in tox to define common baspython as py3 os that\nall tox env including [testenv:functional] will use py3 instead of py2.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: I238a0342269ce3e71050594425ed2c5876ce9b54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/e2ee23a7b0173c99f66ea34163c37de5dd070a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/e2ee23a7b0173c99f66ea34163c37de5dd070a16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
