)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~715001","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~Ic7b31506177ebb0f4f24eaff4db134aace5c0b1a","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"memcached-refactor","hashtags":[],"change_id":"Ic7b31506177ebb0f4f24eaff4db134aace5c0b1a","subject":"Refactor memcached_servers","status":"MERGED","created":"2020-03-25 16:59:36.000000000","updated":"2020-03-26 17:42:49.000000000","submitted":"2020-03-26 17:41:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715001-1585244471924-1d1e633c","meta_rev_id":"28d643eeaceae7fe913d9e181a2ed9103b6c9d5b","_number":715001,"virtual_id_number":715001,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-26 17:41:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-03-26 14:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-26 15:24:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-26 15:24:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 18:05:56.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-03-26 14:07:07.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-03-26 15:24:09.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-26 17:41:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"227a047e3446cebd42f3934183656e02bcbcf6bc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-25 16:59:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb3f4a52ce0ee559ce9c20f8bff8143b900f4e46","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-25 17:12:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e4b39d654e955b89cb0266069e4b03005a3056a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 18:22:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/533cbf244dcd474189f9b7e3d2687646 : SUCCESS in 11m 56s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b746556f277341a8a9eeb4698342b19a : FAILURE in 1h 13m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2248508fb6344966869d64328884885f : FAILURE in 1h 17m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d17150531fa4cbbab367f13d71c6c34 : FAILURE in 1h 07m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260883d67a3641cbbc3596a27a5829d1 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9e305bd78514e2a61017fe82cc257ab118415dde","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-25 20:26:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8eb4006c9243ce926ebc2fdd3508ba98f877aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 22:20:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f14d3c11703b4c0f928e618905f42994 : SUCCESS in 7m 09s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2da58c05fe8b488c9e749c36e9c42616 : SUCCESS in 1h 01m 02s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/9b41b30ebd994979b8d5965c1e6677c7 : SUCCESS in 1h 33m 39s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e543148de4649c8a52966edca4ade55 : SUCCESS in 53m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e46f9a5404477eaaf652948b64d18a : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"07b746a8be20249bc0b3c60c9b8a680257eda5bf","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-26 14:07:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"46770db11d64ddc0b7eae66877fdf025ce535e05","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-26 15:24:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"04adb49fc5ad54baa8ce28bd444602ff46e3d57d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 15:24:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec91f58986e127d051b557605b1aad3905043649","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 17:41:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b4125eff0cf4475db9f9ec833d52ca66 : SUCCESS in 9m 26s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a6fc35c13cb04117a245fab9a0bd8f5e : SUCCESS in 1h 07m 25s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2150251a3c5248129fee8cb4bda324ee : SUCCESS in 1h 15m 27s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac4e14610129428f841ce3f427ee1d58 : SUCCESS in 56m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c6eb318f83450c9dbfb87a5b8e1618 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e805d79e8fcafce06ed622717a8a493767a679db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 17:41:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e185ed31393bd85592afa20e67240d4628e986d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 17:42:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f648fa36484b319b5b3fdbcefde63f : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ee1939c1cd6e7850817988d6497438c3d0dde10","revisions":{"0edd367730e355009debd538a5f52ae347a0bd45":{"kind":"REWORK","_number":1,"created":"2020-03-25 16:59:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/01/715001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/01/715001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/01/715001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/01/715001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/01/715001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/01/715001/1"}}},"commit":{"parents":[{"commit":"399db8a94e5dd62c1ba3f7f08e183e4405666517","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/399db8a94e5dd62c1ba3f7f08e183e4405666517"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-03-25 16:59:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-03-25 16:59:19.000000000","tz":120},"subject":"Refactor memcached_servers","message":"Refactor memcached_servers\n\nThis patch aims to add a prefix for memcached_server\non each role to give the ability for deployers to\noverride the location of memcached cluster. I.e users\nwants to create a single memcached cluster with k8s\nfor each service.\n\nWe also add pymemcache based on [1]\n\n[1] https://review.opendev.org/711429\n\nChange-Id: Ic7b31506177ebb0f4f24eaff4db134aace5c0b1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/0edd367730e355009debd538a5f52ae347a0bd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/0edd367730e355009debd538a5f52ae347a0bd45"}]},"branch":"refs/heads/master"},"7ee1939c1cd6e7850817988d6497438c3d0dde10":{"kind":"REWORK","_number":2,"created":"2020-03-25 20:26:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/01/715001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/01/715001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/01/715001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/01/715001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/01/715001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/01/715001/2"}}},"commit":{"parents":[{"commit":"399db8a94e5dd62c1ba3f7f08e183e4405666517","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/399db8a94e5dd62c1ba3f7f08e183e4405666517"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-03-25 16:59:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-03-25 20:26:05.000000000","tz":120},"subject":"Refactor memcached_servers","message":"Refactor memcached_servers\n\nThis patch aims to add a prefix for memcached_server\non each role to give the ability for deployers to\noverride the location of memcached cluster. I.e users\nwants to create a single memcached cluster with k8s\nfor each service.\n\nWe also add pymemcache based on [1] and fix zun-docker\nsystemd config.\n\n[1] https://review.opendev.org/711429\n\nChange-Id: Ic7b31506177ebb0f4f24eaff4db134aace5c0b1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/7ee1939c1cd6e7850817988d6497438c3d0dde10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/7ee1939c1cd6e7850817988d6497438c3d0dde10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
