)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~734213","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~Id4bf83b08b6324678ab34817d4e64b44281f4a35","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Id4bf83b08b6324678ab34817d4e64b44281f4a35","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-06-08 19:57:42.000000000","updated":"2020-06-17 23:22:50.000000000","submitted":"2020-06-17 23:21:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734213-1592436062184-9d6bb993","meta_rev_id":"5eb5b573f864a8fa610dd8e88feac70cbd652f96","_number":734213,"virtual_id_number":734213,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 23:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-09 05:54:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-09 05:54:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 05:54:27.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-17 23:21:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70867659061ea76a4bb854cd43e049c85c1e1a30","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-08 19:57:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf7e5622c5e445638680b7b5c35412858ee120af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 21:59:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d8fcf4697d214fb28721f84f82992278 : SUCCESS in 11m 06s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/46b018dbfc7c49429d8bd31c57bb6994 : SUCCESS in 1h 08m 38s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1aa54c61645242ba98884fbf363e904e : SUCCESS in 1h 14m 27s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ebb42705a24d48b8840fc6ddc9c41689 : FAILURE in 12m 18s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6362bd7d58c7498885b383df578a1e56 : SUCCESS in 58m 27s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e36d597a1b094759adb0753039056d2e : FAILURE in 19m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e25dd737baa499e8f97cfd146b3e1c2 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"66795e18ed69a29ead28913b05c0d00726b4e25b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-09 05:54:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1cbecf255961ff6b1c17109b0d79315e754ccc09","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-09 05:54:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ef4f37f3fd6416122491dcbada22c1362197f22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 05:54:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6a676eb8e5ff862ac3cce73904f0376a5e44304","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 07:10:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b98a0a693bf046eb8387afa03c8cb98f : SUCCESS in 9m 48s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/fee498a037dd4b8784c37ead97c4f04f : SUCCESS in 1h 15m 23s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/90528b261a33427f9389273c8d73f43b : FAILURE in 1h 12m 48s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bda3168debe4306ada45999a269393b : SUCCESS in 1h 00m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01dfa822a6e34a8c91c27831f8a9b65b : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3e0ae83b08a2e6d42956d3339c88770cdda3e98b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 19:52:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"08584523cf40aabcbacd7681ae9bdcbb88a312fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 21:54:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a3adaa23497d41c189694c5fc3052d74 : SUCCESS in 12m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9f042d10b2524e07a00773630a6239f9 : SUCCESS in 1h 17m 43s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/6f9486e96a114a1a9fedc4d350b8f5b6 : SUCCESS in 1h 13m 57s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/edc33a1954f140c996be1cdae40e120e : FAILURE in 14m 07s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/702de6f830b14a4bb41a4b9a8d470fc7 : SUCCESS in 49m 21s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86a08e8a89844478b9ccccb3f2fbae91 : FAILURE in 34m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b304366194d4d6e9a8305c6cbdb67b8 : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":1},{"id":"71fe704a4d235a880b697d9a84ea572652b81547","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 21:55:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c6a5be9531d0ae289d5029034287fdf2b7f4d4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:21:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6401aacba5e74368ad515be8dfb93a04 : SUCCESS in 14m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a060128a66e94a729eb4b223d8ba9cf0 : SUCCESS in 1h 22m 44s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b5ed37eeefcf42eab5edc64e9893d46e : SUCCESS in 1h 08m 48s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f227c8252f14a2fa25b0b8bbb99ec60 : SUCCESS in 1h 03m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9088509b364d34b7289c9601e69b1f : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e1a35cf915fc49328e544accebf1bd6279871e67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:21:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2b6812b4d71c31df005069eca22084a84e836f2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:22:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a75f29ca113427caca7741128c50d52 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc3de56005fb9a69a5cadb2313f3a8518d857bb0","revisions":{"bc3de56005fb9a69a5cadb2313f3a8518d857bb0":{"kind":"REWORK","_number":1,"created":"2020-06-08 19:57:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/734213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/13/734213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/13/734213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/13/734213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/13/734213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/13/734213/1"}}},"commit":{"parents":[{"commit":"f897983093fbb641732d13b9623075c3949b3a65","subject":"Merge \"Use newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/f897983093fbb641732d13b9623075c3949b3a65"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-08 19:57:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-08 19:57:39.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Id4bf83b08b6324678ab34817d4e64b44281f4a35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/bc3de56005fb9a69a5cadb2313f3a8518d857bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/bc3de56005fb9a69a5cadb2313f3a8518d857bb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
