)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~747255","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~Ic57de36d5d240e6a5dda6e1794aa04d1f77fb962","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"bind-to-mgmt","hashtags":[],"change_id":"Ic57de36d5d240e6a5dda6e1794aa04d1f77fb962","subject":"Use the utility host for db setup tasks","status":"MERGED","created":"2020-08-20 16:51:54.000000000","updated":"2020-08-21 08:28:40.000000000","submitted":"2020-08-21 08:26:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747255-1597998394702-ea02c699","meta_rev_id":"0e1fd3d92ce3b46511fab71f050f8b438914eb72","_number":747255,"virtual_id_number":747255,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 08:26:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":2,"date":"2020-08-20 21:50:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-21 07:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-21 07:16:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 21:50:56.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2020-08-21 07:16:27.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-21 08:26:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42704308e3c090b43167e0b271313004cd2d278a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-20 16:51:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ef88cd3d9eadc8d5c17e63bfd6fa279cef42d71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 18:38:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/046ffab7403d45d89d1e457f07245811 : SUCCESS in 6m 28s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f447b0dc3fd74eb28edd90652f0bc499 : SUCCESS in 1h 00m 32s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/f908dcc2baa849608658a986cb0e390b : SUCCESS in 1h 24m 47s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b2e13d8a3dc34b9aa1a05b09a90bf57e : FAILURE in 50m 56s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf15f17263f1451da070d862d0248829 : SUCCESS in 53m 46s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fdf076aa42b438cb68c91d5aa3957e8 : FAILURE in 35m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b421f1b145a1492e8be15a2a93e7c37b : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7c7abe13a5f8d49b23bd2a2e9c785118b63c9ec3","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2020-08-20 21:50:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f66f399daa75cb61d4d680164ddc39641d73bfc4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-21 07:16:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d038c54eaba087edaf84a9634778450231e27ecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 07:16:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c15931e6a2bb2fb4c0e9b27d28d310300e9dfa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 08:26:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0959a9d62114fd77f9ffd5e3a4e1289a40529161","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 08:26:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0e49c067c3d148e2aff0d72a1ae4d7b2 : SUCCESS in 7m 57s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a297921b98674b53b26a659310c1b3f1 : SUCCESS in 1h 01m 42s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8b67d7a2d814493ca36b43c6c9d64e16 : SUCCESS in 1h 00m 53s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0889580738d74276ba243fac73a29712 : SUCCESS in 54m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/609d15e07d8a4f24b8175f6f2a7374df : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2376c3d1daaa5774fe1fdbe0d873c8815f5cbd79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 08:28:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785cc9f5a04b4a3d878b1f6f1bdd35a0 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07d78458207156378406db2f8cb7a6140c047b47","revisions":{"07d78458207156378406db2f8cb7a6140c047b47":{"kind":"REWORK","_number":1,"created":"2020-08-20 16:51:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/747255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/55/747255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/747255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/747255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/747255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/747255/1"}}},"commit":{"parents":[{"commit":"767a11450cf1acfed5db3bbcb771b98bfd3bca02","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/767a11450cf1acfed5db3bbcb771b98bfd3bca02"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-20 16:51:42.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-20 16:51:42.000000000","tz":180},"subject":"Use the utility host for db setup tasks","message":"Use the utility host for db setup tasks\n\nMove it to the service setup host (defaults to utility[0]) instead\nof the galera[0] host, and use galera_address (defaults to internal VIP)\nas the endpoint instead of a local connection on the db host.\n\nChange-Id: Ic57de36d5d240e6a5dda6e1794aa04d1f77fb962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/07d78458207156378406db2f8cb7a6140c047b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/07d78458207156378406db2f8cb7a6140c047b47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
