)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~754185","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~I14168ae45eb19d4fca10f9803ca559332abf7ec8","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I14168ae45eb19d4fca10f9803ca559332abf7ec8","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-09-24 16:58:56.000000000","updated":"2021-01-04 14:42:39.000000000","submitted":"2021-01-04 14:41:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754185-openstack/openstack-ansible-tests/sync-tests","meta_rev_id":"a90aa749c5d66c954c45a50eea33ed0a8f3efe2d","_number":754185,"virtual_id_number":754185,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-04 14:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-01-04 12:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-01-04 12:54:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 19:28:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-29 11:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 12:54:24.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"b2a440f39fdb7b13ca4f0558f0fb72968c00742c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-24 16:58:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6691fc089ea9c79576631da331bbe1c2442b571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 18:52:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/90d0739e1dd342feb9682e3185bb54b3 : SUCCESS in 7m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f3255024e3e341848f5b06f4d0687dce : SUCCESS in 59m 27s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a84c33ebd3c1422fbaed2c3bd53b61a2 : FAILURE in 49m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05e295da76d1444eb544b9f747a22851 : SUCCESS in 1h 03m 43s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8883359d92f44cc6b8dd57ec66386098 : FAILURE in 30m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333b74004f2644b48439eb807f451e01 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5f32404965a86c9477e8953a197232cc1a3be997","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-01 14:32:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b756b17f9a1f15916b5c5235dc5ec3ae40ca39a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 16:05:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/619a4c262f50493ea401d334c736894d : FAILURE in 8m 25s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/40d239352a9143ab8836d700b5d6af40 : SUCCESS in 1h 02m 44s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/30a25d640ce64867b6dcec42a8319670 : FAILURE in 51m 26s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16cab75fbd834f0cbbcb5e5b66a4190e : FAILURE in 47m 40s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f66b3c0580724a28aea1a618d6ff7a84 : FAILURE in 40m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02acab39dc540fab2ffd1d9725b5a5c : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"d9c4786367313a193e62fa2539e088b03439a118","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 12:23:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2713bf6d9913e61164b1886c9336752957566d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 13:42:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fdb588e0a2db446cac5a7c72f89c8df6 : FAILURE in 10m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8bf66d6e5e2b429ca5397ed496d40138 : SUCCESS in 58m 46s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/f43815facc2d43a09b968ba1ff582873 : FAILURE in 48m 10s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9879ff301aa242b68b9cb968a0afb406 : SUCCESS in 1h 00m 41s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e82561dfcae94cb589a1b61151556e17 : FAILURE in 35m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/966abde0d1a94330a3bf10450ff3b8ec : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":2},{"id":"03586332e6b53549e3db87449117640b448fae88","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-05 19:28:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fe6a2d589b73972d1f547ad4ec7e21c4d4fc9cd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 20:56:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7a850b92f3a4411b97c7a6b8a02816a8 : FAILURE in 10m 49s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/196a533d98a240e98a0fb06b87e89e51 : SUCCESS in 1h 03m 31s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/6c9506aa31834feab0268f494d5388bb : FAILURE in 50m 47s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdf7dbeaacf44137a3d19472df5879ea : SUCCESS in 1h 01m 11s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a38aef1bf80346189819f020321b2f68 : FAILURE in 38m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c0bb1e94ac84b699f0e64757ee4ee79 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"4bd92691af3bd0345c18fad3b96655fee4ad7fcc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-19 09:21:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e40dc8cc2d446117b319269735ab9c2579fb7c77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 11:33:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ed785e6f6f624c01b7adee7c5b9529ad : FAILURE in 12m 31s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/311d6b0eb0864d8886d9e80360f862b7 : SUCCESS in 1h 05m 00s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ca35ed2deb704840b567fb0289e18b91 : FAILURE in 1h 21m 56s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e9e18c7966f4697a924be044a6d01b4 : SUCCESS in 1h 06m 34s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7356c0147ea9416fbc2cd1905160e806 : FAILURE in 37m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/577b3af8ac174d2e8d71b6f1341af17f : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":3},{"id":"edaa215923e011204b22bfd88ccd276d14330baa","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-29 10:09:39.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"62436677aeabd02df2e807ce027bfabb56ee24cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 11:20:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/020e4b8da64e4006bdb248d5092cb9e6 : SUCCESS in 7m 48s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/fbcbec3aeaab44a8a788308ff1581084 : SUCCESS in 58m 55s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/39175d51c75a4ac1aeb8f080e57b0fbb : FAILURE in 48m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3f3302417d54cb9b13d45fd8995e7a7 : SUCCESS in 1h 05m 32s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88bf954ee61047b580d9c3478645f9e4 : FAILURE in 38m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d115da7b778e4afeb3321484ae8b6903 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":4},{"id":"6a44e6b68ff73afb9aad45b8bd45f7e59f126ebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-04 08:28:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"f97b6ee9f6d52f4fa678ae1c6d30f500a585cae6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-04 10:17:40.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"aa9822b2062922eaba96e04c5959b11e3d57b143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 12:28:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/96728812aac74f528165472de1e4338e : SUCCESS in 2h 01m 08s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/57483bef5d724141ad99e2c61ae70e73 : SUCCESS in 2h 01m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f26263e842074c19a1c2d46979f44857 : SUCCESS in 1h 45m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7027a8480ec74080bcacb4bd74bc69f8 : SUCCESS in 1h 29m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/008e85a368124d4e859ed5481ec7c490 : FAILURE in 1h 02m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9901cd7d6c04b0e8637b8d89fedee10 : FAILURE in 48m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c6b89af86384e458f95811bcda6a0f2 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":6},{"id":"4c2462db5e3eb29418def0974438ee822a454687","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-04 12:54:24.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1b4a618cd4b42079cb895104b3304df9faf0d336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 12:54:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f8d0a15319e6efb23de9ebec3f1443f041127f8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 14:41:24.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6498b9184120418d9836e0587db41a52 : SUCCESS in 1h 31m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/643ead2bf6eb4795b5a469cac4d3541b : SUCCESS in 1h 45m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1714d4dcdf554000831f790be7f7178f : SUCCESS in 1h 35m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65fc761a7d52433d8bf8d9fce2214b3c : SUCCESS in 1h 23m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4f553a31b548868d6ca41e059dc653 : SUCCESS in 5m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b929c85bd224996b0450dcf4267d5d3 : FAILURE in 48m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a2b323fbf3e1e46951bb18ffc6e0eb36779f0ac4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 14:41:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"a90aa749c5d66c954c45a50eea33ed0a8f3efe2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 14:42:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32dfd99bb26847fea604817ea975c4c4 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f82044fc5242bd98f7c8f31df86ac156e07f20bf","revisions":{"c1184730ded627e9dfee8f6253a57c3ce456ac1b":{"kind":"REWORK","_number":1,"created":"2020-09-24 16:58:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/754185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/85/754185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/1"}}},"commit":{"parents":[{"commit":"07d78458207156378406db2f8cb7a6140c047b47","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/07d78458207156378406db2f8cb7a6140c047b47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 16:58:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 16:58:54.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I14168ae45eb19d4fca10f9803ca559332abf7ec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/c1184730ded627e9dfee8f6253a57c3ce456ac1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/c1184730ded627e9dfee8f6253a57c3ce456ac1b"}]},"branch":"refs/heads/master"},"dc2c545ed9337472d64eed63f3dcf7f9199953c3":{"kind":"REWORK","_number":2,"created":"2020-10-01 14:32:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/754185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/85/754185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/2"}}},"commit":{"parents":[{"commit":"07d78458207156378406db2f8cb7a6140c047b47","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/07d78458207156378406db2f8cb7a6140c047b47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-01 14:32:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-01 14:32:34.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I14168ae45eb19d4fca10f9803ca559332abf7ec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/dc2c545ed9337472d64eed63f3dcf7f9199953c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/dc2c545ed9337472d64eed63f3dcf7f9199953c3"}]},"branch":"refs/heads/master"},"ab8f6fa38d07a6c06df0004c38d7e7ba9571c5f7":{"kind":"REWORK","_number":3,"created":"2020-10-19 09:21:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/754185/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/85/754185/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/3"}}},"commit":{"parents":[{"commit":"07d78458207156378406db2f8cb7a6140c047b47","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/07d78458207156378406db2f8cb7a6140c047b47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 09:21:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 09:21:14.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I14168ae45eb19d4fca10f9803ca559332abf7ec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/ab8f6fa38d07a6c06df0004c38d7e7ba9571c5f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/ab8f6fa38d07a6c06df0004c38d7e7ba9571c5f7"}]},"branch":"refs/heads/master"},"1916d683011573f7787b82746615e9e7ce29dd4b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-10-29 10:09:39.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/85/754185/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/85/754185/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/4"}}},"commit":{"parents":[{"commit":"07d78458207156378406db2f8cb7a6140c047b47","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/07d78458207156378406db2f8cb7a6140c047b47"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 09:21:14.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2020-10-29 09:10:08.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nDepends-On: https://review.opendev.org/760304\nChange-Id: I14168ae45eb19d4fca10f9803ca559332abf7ec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/1916d683011573f7787b82746615e9e7ce29dd4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/1916d683011573f7787b82746615e9e7ce29dd4b"}]},"branch":"refs/heads/master"},"df343988c034ce1b00da7c9826f7dc6e17c17157":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-04 08:28:42.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/85/754185/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/85/754185/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/5"}}},"commit":{"parents":[{"commit":"11a5ce29910ff85efb378e81d71e4e37f33b8da5","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/11a5ce29910ff85efb378e81d71e4e37f33b8da5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 09:21:14.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-04 08:28:42.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nDepends-On: https://review.opendev.org/760304\nChange-Id: I14168ae45eb19d4fca10f9803ca559332abf7ec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/df343988c034ce1b00da7c9826f7dc6e17c17157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/df343988c034ce1b00da7c9826f7dc6e17c17157"}]},"branch":"refs/heads/master","description":"Rebase"},"f82044fc5242bd98f7c8f31df86ac156e07f20bf":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-01-04 10:17:40.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/85/754185/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/85/754185/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/85/754185/6"}}},"commit":{"parents":[{"commit":"11a5ce29910ff85efb378e81d71e4e37f33b8da5","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/11a5ce29910ff85efb378e81d71e4e37f33b8da5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 09:21:14.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-04 10:17:40.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I14168ae45eb19d4fca10f9803ca559332abf7ec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/f82044fc5242bd98f7c8f31df86ac156e07f20bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/f82044fc5242bd98f7c8f31df86ac156e07f20bf"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
