)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~769116","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~Ie5c741d47d114222934ad01097710fa8dc56dd4c","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"zun-nova","hashtags":[],"change_id":"Ie5c741d47d114222934ad01097710fa8dc56dd4c","subject":"zun: set configuration where compute host shared with nova","status":"MERGED","created":"2021-01-04 09:33:52.000000000","updated":"2021-01-10 16:36:48.000000000","submitted":"2021-01-10 16:35:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769116-zun-nova","meta_rev_id":"79733bb455279c7e104c489cd18ca9969a713130","_number":769116,"virtual_id_number":769116,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-10 16:35:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-01-10 16:36:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-10 16:36:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2021-01-08 11:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-04 11:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-05 08:36:43.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-01-05 21:12:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"a803de730dfe3f19c9e5d69364c8504c8d6acf62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-04 09:33:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"517ab42c1d9de39868f333bac37bcbce2f93b86c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 11:02:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ae0b2d05ebf2400784978381564a0805 : FAILURE in 55m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e4e10bfced844ff2aa59e2569cc61833 : FAILURE in 1h 19m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b95e5e1d99064794997ff52905a7d4cf : FAILURE in 1h 11m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5214ad96892445ceaa44a2764622297b : FAILURE in 1h 06m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e9b9c383056e49eb9932e41f54f18a44 : FAILURE in 1h 03m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6ae4f437f5b44d39833ad3bf44b2e70 : FAILURE in 47m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7de7b34f8b43bca7410a76eb61f4b1 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ddce258d3914b0284818e2ed29f025c0bf728e90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-04 13:20:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76b13e76e38a234b7f23763746a8839be5aa5e20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:17:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d85e220107464bdb942d18ef2dd01c9b : SUCCESS in 1h 19m 52s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/04dbe97d79be4eafb887856652049f66 : SUCCESS in 1h 50m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc1cd321808b4b55a7ffdabe59d2b769 : SUCCESS in 1h 39m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49cfdfcfda7a4c8690e14ba9bb96d1fa : SUCCESS in 1h 29m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e6597d7982d7404c90660139a1ef04b8 : FAILURE in 1h 00m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5759b90462b7491683abe75dc837da66 : FAILURE in 52m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b8e851b5b248b28b199c6a25cd9545 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a62a646df50d9df3bd203429fbfbacc26bd64462","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-05 08:36:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eaa486a88d7a3d9d0a19e8858a415c5acaf18f0d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-05 21:12:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"674198e4456acaf238b85b96fd4204778271b6ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 21:12:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c3718eefb6c511555d5f1bdc2fd95ca09de70776","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 23:45:29.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4934795b461449548ce21e0419b2efdd : SUCCESS in 1h 40m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7c8ef4f32db04846a9d068ad99c6a330 : SUCCESS in 1h 53m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cc4884fd4dc407882b958c69b297c6b : SUCCESS in 1h 42m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d40bf2b4187b49c0a6196bfd2ff58a86 : FAILURE in 1h 34m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c4277b90e14865be598119cd0030ae : SUCCESS in 11m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/640d3e620c814e74afeefd0d0703e888 : FAILURE in 1h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ba6f070d6c301fbd8e494367643c99446f71173","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-06 08:25:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8359633667b7691d5cb3e39079b64a089bcc000e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 11:38:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9240403a8cc841de85779e863826e9be : SUCCESS in 1h 26m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/183c7e496a384e2e9a2b89780b814f50 : FAILURE in 1h 50m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e791e050ca78485c86ef572911651049 : SUCCESS in 2h 06m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a27a1f416594e7f93bdc0d8e8d970d7 : TIMED_OUT in 3h 05m 38s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8fd473d8d0cc4329b7bfbfdae2dbdb6c : FAILURE in 1h 00m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dbb6a14974a44b2abb61e8abd7c8c8f : FAILURE in 55m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de69fb7525f64b8eaba12460bdf07cdd : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4141de7bb1d2adf8c3fcbfc781eafd899602e9d2","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 11:21:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"436376aafa85b4ecd551a4e0c958fbe21076bf95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 13:02:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/30f75e199a7f4bba8c3fce96fbd58502 : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c78cbd6e2e044fe48ef9013e4fa2400d : SUCCESS in 1h 32m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/401e21ee589848eda68c9cc82d1acd7f : SUCCESS in 1h 35m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd1b51b495784b36924cc865da27ae9c : FAILURE in 1h 31m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6c5e602cdaf3482f835887d53528beb9 : FAILURE in 1h 12m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5b52f9aa0c1430a8dfc39d041c83459 : FAILURE in 51m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db26b176d8df41b1b47cb28d0ad601aa : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cfc4a994c32b432fc72765b5bd2978c72ed37b62","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 13:08:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f7b4eed7447acd5c304ae06c8895bf6e53af43bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 15:19:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b7092d82d71940ea91e7f2905486d06e : SUCCESS in 1h 26m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c8184d7618cd4fbbb56635d3d6ffa0fc : SUCCESS in 1h 33m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e6d1c0e29864f6d963e779773f0faa8 : FAILURE in 1h 59m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f293a1ba49ac45f79fc9ba8d1a28d27b : SUCCESS in 1h 22m 57s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0b906aa9a45e4b75a527c15f762905cc : FAILURE in 1h 08m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d99d502271e4e7889e62e6450e8a7cf : FAILURE in 55m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48910d3506b6494987325c981e543ef2 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"26dabb5dc985023f220b67d2f5182eb6e0fd42bd","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 16:05:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7923655638e30da108810a54d670b47ca06d522c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 19:13:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bff3ff87309e4300a98958b6b3531109 : SUCCESS in 1h 45m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c0c820e8d35140bdb06bc3c28ba3689e : SUCCESS in 2h 05m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f44fa9afe30247839c7e46f374677e95 : TIMED_OUT in 3h 04m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d93b0516f654054a6c86ec3fcebde39 : TIMED_OUT in 3h 04m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/72f2a2772d1c41bb9c8ff8fcbdcaaef6 : FAILURE in 1h 07m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/210638e638a84967967b715f18931f60 : FAILURE in 57m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc3a1abb2634d7ca56cef1cb5acb040 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"885315c73b2b4bdf9ea2642e66a26657e19dd3db","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 21:54:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f4f669980f4c6bae835282ab5780e0777e8efb4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 23:52:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/986cfed68e374e81b9219cfedc08171b : SUCCESS in 1h 32m 49s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b99189e00d204dfe9166937655519322 : FAILURE in 1h 51m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb4e1fb09e5741cdb0fe5d20718fcf50 : SUCCESS in 1h 37m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b692edcab2ec4d309d90c08d2280e0a1 : SUCCESS in 1h 27m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c1f69f784c6b46f3846dcf1c15093350 : FAILURE in 1h 27m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b58e5987f10446a98ea89e29d5d95d16 : FAILURE in 51m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171e90eaca334dc694b7884ae11d8cbe : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":2},{"id":"3d2d11ba264b8cc3a2f05ce3473077a1422913ae","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-08 08:31:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"03c0ba4784f85b36f6173eb373559f31259415dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 10:38:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/069809b50f8147dd8ab5911adb3a7bf8 : FAILURE in 1h 59m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a81e25df29d3426c8dc93b147c774b6c : SUCCESS in 1h 55m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/985d0b34b67649bb86694ec7d22bd64f : SUCCESS in 1h 32m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caa498ce554f40b688e42187e830b171 : SUCCESS in 1h 28m 01s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ea495a48bb664ce09b5e92f686efc11e : FAILURE in 1h 00m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a10d2c7cd66482fad5fee8a0e6a245f : FAILURE in 54m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed212b222e0841d28ab5e421763fbcfb : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"0bcfe10992b627f653c871e28991f363be0cbedd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-08 11:39:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"41004d6a985abc9b984c89ae126f4a7120e8a3d5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-08 11:39:34.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"26c7ca75533af933edf92f1fd3ea0bb014061247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 13:37:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b4634de95cb9457184c5c96bb2ffb999 : SUCCESS in 1h 52m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/82a9a0db2ea6476797ef51a7514c6714 : FAILURE in 1h 40m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f429d3ce4774a57a54aaa2525d35225 : SUCCESS in 1h 36m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28abd34574f44362b5eaf1866125c138 : SUCCESS in 1h 41m 35s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f021b2af2d854a639fcf5c55f54d70f0 : FAILURE in 1h 11m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a3d43d6ce6c4c6e98e2a3d780dd85bc : FAILURE in 50m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c24bb51e69d94a54ac19ffa7f629b96d : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":3},{"id":"1806a4cb7a5af4fa7afda10f967dd69232cf64be","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-08 13:39:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"285289833410c056f0c5009797e424a7d8885b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 15:57:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/024036aaf3be4f0ea4621b2c106f1072 : SUCCESS in 1h 36m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/79043968c354484bb80c80a01a172bb1 : FAILURE in 2h 14m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a80ad3c8ff9445df9d6e9f7621c66c7a : SUCCESS in 1h 35m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea249c23a78a4916a4c6f08f3a8e1c29 : SUCCESS in 1h 27m 01s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3b9a6ab11e7a4480b6309472e09243a0 : FAILURE in 1h 06m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da85dd58840742dab7b3d1d73f195f2d : FAILURE in 51m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa00adb4fd1a431ab7c483a30f7ec740 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"9098f27e7022dedf6aefd65c942e20b11241042a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-08 18:10:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"565053f3dee35e5bdf4fc53dc859e27f538ac87f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 20:19:54.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/683a71df39e5446583f2e66b3f61fa1d : SUCCESS in 1h 40m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a1fa99d989f45339754ecf17021a5f1 : SUCCESS in 1h 52m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/059b04aba34d4efe973c6a842c59a638 : FAILURE in 1h 47m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2c3e062c8c440438b7cf13379b6ae33 : SUCCESS in 1h 23m 56s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/846ef6bff2a24b1889c7106553128647 : FAILURE in 1h 03m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9132ed67887c401199fac043eb224788 : FAILURE in 52m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/748cff806d614aed80590b20998f30ed : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"af9e012863fed2ffbe993c9c289ee29f1527f817","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-09 11:01:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"287e4ed7665cd11d254a6d553d226a9e9bcd443c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 13:08:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/77daa82403854f7e9c2b0a3edfd912ee : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ad946dc4a70048edab0b93e40eef9c68 : SUCCESS in 1h 56m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8eca221bb3274a55bf9343661608a264 : SUCCESS in 1h 32m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/949191bcdeec471889de90329781c622 : SUCCESS in 2h 00m 50s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/56804acd2edc4654ad983b2ccfa4ba25 : FAILURE in 55m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86abb97b984f4b59a1ead9a70d93d686 : FAILURE in 54m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7468648f2742468d95689a082f43fcbc : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8d40eac6564a3846f4b6f03ec6986166f7386fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 13:08:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2610028ee963e945b540dc2c8921b4a7a0756367","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 14:52:34.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f3062715a52444efbb510f17afbd3f2d : FAILURE in 1h 38m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/194c40674196469588b64c0187c518e9 : SUCCESS in 1h 41m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/073d02a4212949ceb69563757b2f0312 : FAILURE in 1h 29m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ab460dbef1c46a885f4d27b7689b8ab : FAILURE in 1h 29m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913c116145294bfaa93fabdb55b4029a : SUCCESS in 5m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da31dfa787a84888bbdf1c6ea4f9cb3b : FAILURE in 1h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92d89b0898f13a0de2c2020eeddc4df3c8f10f99","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-09 16:20:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"60914db1e8b9f8d80ae56b3c0235a598f2819a72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 19:19:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b9672e9589ea4dfda6b0d097c706fb32 : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f4cc0aab21574a139d90afba030d8eb0 : FAILURE in 2h 56m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71eb07dd989d443d8995abff46b53dd7 : SUCCESS in 1h 42m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dfeec2e7b394606b0c5e5ed75658eb8 : SUCCESS in 1h 41m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7f17fc07f37d4211ac65a44a14b837b6 : FAILURE in 1h 05m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dccccaa8405d40448e983e06a4f0f971 : FAILURE in 50m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb149713d4843a580128b754d8a91e9 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"a4e1053be59c2a4bad935243bc80da5c7fd72ba5","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-10 13:10:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"866536a670bd6b13312021758a63348afa7e2b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-10 14:47:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0aad21b70480471b87e599b1bb46f371 : SUCCESS in 1h 27m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f0b6761dbf654d77ace8928f8eaf2ac7 : SUCCESS in 1h 31m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89394d13f9444e538e1d0a74144533f5 : SUCCESS in 1h 34m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80f14893480744eaac14148f3f6cd6f5 : SUCCESS in 1h 18m 13s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ab61a717b21c4195a64157ee37c50c26 : FAILURE in 1h 03m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/358e3edc956647ccbff1ca847227202a : FAILURE in 55m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77ebe0952864a72b8af00193f01baa6 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":3},{"id":"fdd884da12fe4df0b52ab8c8b0be825ed1cb60f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-10 14:48:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a5fad9b62c2aa705196ca29cc08e7da07576bf8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-10 16:35:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8c97f1b48f4e4ad4b3a3b964cfcfd7f3 : SUCCESS in 1h 26m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/81522c6a87154bafb505b448066af4b4 : SUCCESS in 1h 41m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cd3d99cbac2416a91b3c1eaf6c41484 : SUCCESS in 1h 36m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cac1b707eab54fe6b0bbd9e1deb00c88 : SUCCESS in 1h 29m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1036f07e8ae4ad99a9c5166a0519737 : SUCCESS in 4m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74cb40a85d8d42d68cb11e3ec8afc720 : FAILURE in 52m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0b86a2b8e32e5d90534361c2df91c0dee13eb2e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-10 16:35:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5d40551b769b242b857fd1fb3f5044344c0a594c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-10 16:36:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9fc5ae34c4d4ca78f8ce52302ab93f9 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"83c4435950f9ee01c863f36315206dcc57050dad","revisions":{"4a61fbdaab80100eb7fc399153a2bc0d00649e43":{"kind":"REWORK","_number":1,"created":"2021-01-04 09:33:52.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/16/769116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/16/769116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/1"}}},"commit":{"parents":[{"commit":"11a5ce29910ff85efb378e81d71e4e37f33b8da5","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/11a5ce29910ff85efb378e81d71e4e37f33b8da5"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2020-12-02 15:54:58.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-04 09:26:12.000000000","tz":0},"subject":"zun: set configuration where compute host shared with nova","message":"zun: set configuration where compute host shared with nova\n\nThis ensures that only one record is generated in placement\nfor each compute host when both Nova and Zun run alongside\neach other.\n\nChange-Id: Ie5c741d47d114222934ad01097710fa8dc56dd4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/4a61fbdaab80100eb7fc399153a2bc0d00649e43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/4a61fbdaab80100eb7fc399153a2bc0d00649e43"}]},"branch":"refs/heads/master"},"c895c98a17c63b80f0345169962fe2631496d483":{"kind":"REWORK","_number":2,"created":"2021-01-04 13:20:34.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/16/769116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/16/769116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/2"}}},"commit":{"parents":[{"commit":"11a5ce29910ff85efb378e81d71e4e37f33b8da5","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/11a5ce29910ff85efb378e81d71e4e37f33b8da5"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2020-12-02 15:54:58.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-04 13:19:55.000000000","tz":0},"subject":"zun: set configuration where compute host shared with nova","message":"zun: set configuration where compute host shared with nova\n\nThis ensures that only one record is generated in placement\nfor each compute host when both Nova and Zun run alongside\neach other.\n\nChange-Id: Ie5c741d47d114222934ad01097710fa8dc56dd4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/c895c98a17c63b80f0345169962fe2631496d483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/c895c98a17c63b80f0345169962fe2631496d483"}]},"branch":"refs/heads/master"},"83c4435950f9ee01c863f36315206dcc57050dad":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-08 11:39:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/769116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/16/769116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/16/769116/3"}}},"commit":{"parents":[{"commit":"a44292615808c4b42938df0475d63bcd272fe1af","subject":"Merge \"defaults: start kuryr service before docker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/a44292615808c4b42938df0475d63bcd272fe1af"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2020-12-02 15:54:58.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-01-08 11:39:29.000000000","tz":0},"subject":"zun: set configuration where compute host shared with nova","message":"zun: set configuration where compute host shared with nova\n\nThis ensures that only one record is generated in placement\nfor each compute host when both Nova and Zun run alongside\neach other.\n\nChange-Id: Ie5c741d47d114222934ad01097710fa8dc56dd4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/83c4435950f9ee01c863f36315206dcc57050dad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/83c4435950f9ee01c863f36315206dcc57050dad"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
