)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~769555","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~Iea07012d092333c656b397f97b541a2f0a5f0e44","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"zun-kata","hashtags":[],"change_id":"Iea07012d092333c656b397f97b541a2f0a5f0e44","subject":"Add support for kata container runtime","status":"MERGED","created":"2021-01-06 15:26:23.000000000","updated":"2021-01-18 16:15:55.000000000","submitted":"2021-01-18 16:11:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"769555-zun-kata","meta_rev_id":"f288fa4e7b5dac1bcf744e43c1681f0891441575","_number":769555,"virtual_id_number":769555,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 16:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-01-18 14:03:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-14 08:34:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2021-01-18 14:03:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-06 16:47:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-11 12:31:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-01-18 14:03:03.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"e425d0310b3790c19597956cc7c269141220aad8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-06 15:26:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8351ccd1174512f19d52740ba1d88d3ed657936a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-06 15:26:37.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3cbeb8c4400c0cc03d1881d39fe5165ccb36dbf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 16:47:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e9d72b03eb33436da798bd3ee5a699a2 : FAILURE in 23m 29s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/28db829860714193b86c1f659ca07903 : FAILURE in 27m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32b44fab103a4d04a334b1781cc01d43 : FAILURE in 26m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9a0e953aba74478aa1ffe8d43b081f9 : FAILURE in 24m 14s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f077b9d9d879404e8cf3964d0f96eb3d : FAILURE in 1h 03m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8ab8f575fc84bfc9bb50bc17bf2f75c : FAILURE in 57m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e2e7e31852415d9a78e74b71e13878 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7e78bd11342307cdf0d3873f1abf0476e9078d06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 08:29:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85e81215e665d5beef8e7c55c90b2be0e17f2b01","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 08:29:42.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"08d057f9babc261424003bac7af6ca6e945687ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 10:01:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c6598f202e244abba5836b92f3509b5c : FAILURE in 1h 23m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b4b9a8f347d84289925c9e26461cb1f9 : FAILURE in 1h 14m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7f5cf1128fa4071b70612179c6fcbcf : FAILURE in 1h 03m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43bfa4b869e4457eb87cc6458671d43c : FAILURE in 1h 20m 52s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e07f80dc2d82450b879bfba12f2214cd : FAILURE in 1h 06m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5046aa37a4444dc29ad0f828a256893d : FAILURE in 42m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1916c3a224984ef1a3de23b19d5b729f : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2fc758adeccddca5b3c6d45d39d021316ff3e175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 10:48:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d1a45861e61def461f9f9aec34ee3d9258a9b48","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 10:48:38.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"c68e5256f0fbe62c96fc089ced9bb8f5c0a08faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 12:45:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/219a3abf5485477f952c7d23ed603ddb : FAILURE in 1h 13m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5cf4e111b2b0467182fdac890f24f4a6 : SUCCESS in 1h 51m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0a8fc4d36ed4c35b6f99333a6efe440 : SUCCESS in 1h 44m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8de52469a2b1440c87f094a2e8cd98bc : FAILURE in 1h 37m 27s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4b2ec2cfe9f0405ba6af49811124eba5 : FAILURE in 1h 05m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb99b8aaa5954bebb287fdddb3fc01e5 : FAILURE in 1h 01m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69a60f5ff014e8db640ebc70dc21a5c : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3},{"id":"61d7936892f0536f581c3cdbce1174d63b6a9116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 13:17:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d455c33ab282bf13964b0d99b8458a313d401750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 14:37:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b3e5339e117d7350053f99458e77fcb0f61d2b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 16:42:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1b406c47070e41548329aac90afdc32e : SUCCESS in 1h 40m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4d71e00ebf1345298f53bfbb3df0873c : SUCCESS in 1h 51m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46887fdd44f54d449c1e579ff3990548 : FAILURE in 1h 51m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/697ff0c522e44471aca4d58607b3150c : SUCCESS in 1h 28m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5e1625d513334e0bb580a0abf6c12525 : FAILURE in 1h 01m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/575fadde99a14eb8a6d929b4310e4805 : FAILURE in 51m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188780c039494c6d902460b6bf0f1e42 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":5},{"id":"dd187b83a3f05bb58aa32a087e293f5182c1e6d4","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-07 21:53:30.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ad0d661fa6c4fdc6e706f52f61adcd22b8a2cfd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 23:48:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e0ba6376041244a7b1f8a07acf8035ba : SUCCESS in 1h 25m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/31cc34e6c32b4a3d8cbe889fce5ee666 : SUCCESS in 1h 45m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6c4f51f87ed4d358aacba34b6cc8f7f : FAILURE in 1h 41m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9235407e0f949caa2f90328e0937059 : SUCCESS in 1h 26m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/68438aacc6a54116b5516452e71eb821 : FAILURE in 1h 06m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91a64353119044239e20218c74f82c2c : FAILURE in 53m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fe8aac5d0734de0b1a1dc1db661730e : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":5},{"id":"1353591d7f11982c6f73f51ee56e13c13059e139","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-08 13:30:27.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"58fb6d803bca86c7c0b18d8460f2c8373f435259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 15:54:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/54ec6eab40f24f14b58f2c08b141bbce : SUCCESS in 1h 28m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2af1896ca20844a682d4910586cb4a61 : FAILURE in 2h 19m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4289b69eb0344dc7b3fa4908f7b43467 : SUCCESS in 1h 46m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d85834d1b4f4540bfc9f917af431e5b : FAILURE in 1h 54m 34s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7a0fa24e90774f9ea1ff669bc638ba71 : FAILURE in 1h 09m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2293a9e0846a4aec91b1399db2a72b50 : FAILURE in 59m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cba517db4f74e5781fe195ccd0bf243 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":6},{"id":"e25ff4421ea6ff3f3e3824dab37a11f1957ef392","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-09 13:49:21.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"4585718cb0722a4810d8cf4a2c0ec833c346156e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 15:40:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2ca8f386fc93405fabcb4fdbb1639603 : SUCCESS in 1h 29m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/afa3b6ca04d64fba975c6dd6aacdab5e : SUCCESS in 1h 44m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d36c3c8e3e374b5b818df2b9d12df702 : SUCCESS in 1h 36m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a818d9fd36440429abdaa3e0c0b4fe6 : SUCCESS in 1h 25m 38s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d15f6cb2fd1345e49b250ee4f433fc05 : FAILURE in 53m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4236f892a4b4e7aba3a4b5fb96859f9 : FAILURE in 52m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d0dba5fa8264d2e9f4c02e7dad73f19 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":6},{"id":"034bc8e9d080080ced2956f165a9c7a14aef1d78","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-11 12:31:44.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)\n\nSmall comment otherwise looks good","accounts_in_message":[],"_revision_number":6},{"id":"eb89c308068518bc82c0aaf76b00fd05187dc610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-12 08:32:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cdaf9c9a99fe6b8cd727f971bf819005f8a9c381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 10:20:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/901fe598843541958e72a31bfbf9a1e1 : SUCCESS in 1h 32m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/02d6a4bd3c584bde81056cf12bd53ad2 : SUCCESS in 1h 41m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8891593ab73f4861ad3521c0e17c3df1 : SUCCESS in 1h 34m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9efb01459944f53a277d5d96d7ecdc9 : SUCCESS in 1h 37m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/83e495991e1f43139503f374361e6335 : FAILURE in 1h 10m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a980f9ea3a164882bbc9106c670a504a : FAILURE in 1h 06m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9738b9d4c2848d5b6beaea2befe34d2 : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":7},{"id":"1d98de91820f2e609ce8eb8cdfb459ba0ab1ce46","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-14 08:34:03.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"c477361739086e5869d6d69d8f4cf933bfa1cf79","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-18 14:03:03.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"7be2e7494b29d48912d6b3edfe28443a0fa1e673","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:03:23.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1d598a880b812644613ce61b30f3c1a6870596ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:11:46.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ee2a1e5e4c414ee681fdbef246c792ef : SUCCESS in 1h 33m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bf3008b4bb4e4f1eb2019ae8daed58de : SUCCESS in 1h 54m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b6e644c22ef452b900a936f60e63159 : SUCCESS in 1h 31m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/204d03d13f82482f9ec8c04cb8e8bcdc : SUCCESS in 1h 24m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8454e76fd4df418fa3f2b2725c5aefde : SUCCESS in 5m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07d498b1aca04c86b6010a732e960041 : SUCCESS in 1h 55m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"38a263aeb8f4d62ee3051075c4596bbcf65aea98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:11:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"f288fa4e7b5dac1bcf744e43c1681f0891441575","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:15:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6ededcc40f041abaea330f7727c5fa6 : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"fb440ce3d0d2749d5377266e64b765e6da172443","revisions":{"ae065d39c71de38a6e3554042440490b12bd70b6":{"kind":"REWORK","_number":1,"created":"2021-01-06 15:26:23.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/55/769555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/55/769555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/1"}}},"commit":{"parents":[{"commit":"f82044fc5242bd98f7c8f31df86ac156e07f20bf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/f82044fc5242bd98f7c8f31df86ac156e07f20bf"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-06 14:19:01.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-06 15:25:40.000000000","tz":0},"subject":"Add support for kata container runtime","message":"Add support for kata container runtime\n\nThis adds support for kata containers by installing and configuring\nthe relevant runtime.\n\nThe default remains as \u0027runc\u0027 but can be adjusted using the\nvariable added to the defaults.\n\nChange-Id: Iea07012d092333c656b397f97b541a2f0a5f0e44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/ae065d39c71de38a6e3554042440490b12bd70b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/ae065d39c71de38a6e3554042440490b12bd70b6"}]},"branch":"refs/heads/master"},"6a8796d4396ab0f04a49e11d9f0293bb758e5119":{"kind":"REWORK","_number":2,"created":"2021-01-07 08:29:19.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/55/769555/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/55/769555/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/2"}}},"commit":{"parents":[{"commit":"f82044fc5242bd98f7c8f31df86ac156e07f20bf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/f82044fc5242bd98f7c8f31df86ac156e07f20bf"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-06 14:19:01.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-07 08:28:07.000000000","tz":0},"subject":"Add support for kata container runtime","message":"Add support for kata container runtime\n\nThis adds support for kata containers by installing and configuring\nthe relevant runtime.\n\nThe default remains as \u0027runc\u0027 but can be adjusted using the\nvariable added to the defaults.\n\nChange-Id: Iea07012d092333c656b397f97b541a2f0a5f0e44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/6a8796d4396ab0f04a49e11d9f0293bb758e5119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/6a8796d4396ab0f04a49e11d9f0293bb758e5119"}]},"branch":"refs/heads/master"},"1e561559e9e4d9a2e7b8b5649e435f21cecc9edd":{"kind":"REWORK","_number":3,"created":"2021-01-07 10:48:26.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/55/769555/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/55/769555/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/3"}}},"commit":{"parents":[{"commit":"f82044fc5242bd98f7c8f31df86ac156e07f20bf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/f82044fc5242bd98f7c8f31df86ac156e07f20bf"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-06 14:19:01.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-07 10:48:03.000000000","tz":0},"subject":"Add support for kata container runtime","message":"Add support for kata container runtime\n\nThis adds support for kata containers by installing and configuring\nthe relevant runtime.\n\nThe default remains as \u0027runc\u0027 but can be adjusted using the\nvariable added to the defaults.\n\nChange-Id: Iea07012d092333c656b397f97b541a2f0a5f0e44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/1e561559e9e4d9a2e7b8b5649e435f21cecc9edd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/1e561559e9e4d9a2e7b8b5649e435f21cecc9edd"}]},"branch":"refs/heads/master"},"68f77a1ddb1f8c8a70e4f13ef00b80d8f73b662a":{"kind":"REWORK","_number":4,"created":"2021-01-07 13:17:53.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/55/769555/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/55/769555/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/4"}}},"commit":{"parents":[{"commit":"f82044fc5242bd98f7c8f31df86ac156e07f20bf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/f82044fc5242bd98f7c8f31df86ac156e07f20bf"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-06 14:19:01.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-07 13:17:00.000000000","tz":0},"subject":"Add support for kata container runtime","message":"Add support for kata container runtime\n\nThis adds support for kata containers by installing and configuring\nthe relevant runtime.\n\nThe default remains as \u0027runc\u0027 but can be adjusted using the\nvariable added to the defaults.\n\nChange-Id: Iea07012d092333c656b397f97b541a2f0a5f0e44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/68f77a1ddb1f8c8a70e4f13ef00b80d8f73b662a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/68f77a1ddb1f8c8a70e4f13ef00b80d8f73b662a"}]},"branch":"refs/heads/master"},"42dc19ade492e13995628cbb0784116128800af4":{"kind":"REWORK","_number":5,"created":"2021-01-07 14:37:35.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/55/769555/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/55/769555/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/5"}}},"commit":{"parents":[{"commit":"f82044fc5242bd98f7c8f31df86ac156e07f20bf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/f82044fc5242bd98f7c8f31df86ac156e07f20bf"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-06 14:19:01.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-07 14:37:03.000000000","tz":0},"subject":"Add support for kata container runtime","message":"Add support for kata container runtime\n\nThis adds support for kata containers by installing and configuring\nthe relevant runtime.\n\nThe default remains as \u0027runc\u0027 but can be adjusted using the\nvariable added to the defaults.\n\nChange-Id: Iea07012d092333c656b397f97b541a2f0a5f0e44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/42dc19ade492e13995628cbb0784116128800af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/42dc19ade492e13995628cbb0784116128800af4"}]},"branch":"refs/heads/master"},"61958afa5ac6c1baea0b02ac760bc6a77d032b9c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-01-08 13:30:27.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/55/769555/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/55/769555/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/6"}}},"commit":{"parents":[{"commit":"a44292615808c4b42938df0475d63bcd272fe1af","subject":"Merge \"defaults: start kuryr service before docker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/a44292615808c4b42938df0475d63bcd272fe1af"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-06 14:19:01.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-08 13:30:27.000000000","tz":0},"subject":"Add support for kata container runtime","message":"Add support for kata container runtime\n\nThis adds support for kata containers by installing and configuring\nthe relevant runtime.\n\nThe default remains as \u0027runc\u0027 but can be adjusted using the\nvariable added to the defaults.\n\nChange-Id: Iea07012d092333c656b397f97b541a2f0a5f0e44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/61958afa5ac6c1baea0b02ac760bc6a77d032b9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/61958afa5ac6c1baea0b02ac760bc6a77d032b9c"}]},"branch":"refs/heads/master","description":"Rebase"},"fb440ce3d0d2749d5377266e64b765e6da172443":{"kind":"REWORK","_number":7,"created":"2021-01-12 08:32:04.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/55/769555/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/55/769555/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/55/769555/7"}}},"commit":{"parents":[{"commit":"29a7510b188d263372c1e2e18955be9aa22a1717","subject":"Merge \"Add timed cleanup script to handle old Docker data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/29a7510b188d263372c1e2e18955be9aa22a1717"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-06 14:19:01.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-12 08:30:57.000000000","tz":0},"subject":"Add support for kata container runtime","message":"Add support for kata container runtime\n\nThis adds support for kata containers by installing and configuring\nthe relevant runtime.\n\nThe default remains as \u0027runc\u0027 but can be adjusted using the\nvariable added to the defaults.\n\nChange-Id: Iea07012d092333c656b397f97b541a2f0a5f0e44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/fb440ce3d0d2749d5377266e64b765e6da172443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/fb440ce3d0d2749d5377266e64b765e6da172443"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
