)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~769965","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~Iced1c0986f448b2a19b1f81586e50a039c23f566","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"osa/service_variables","hashtags":[],"change_id":"Iced1c0986f448b2a19b1f81586e50a039c23f566","subject":"Use global service variables","status":"MERGED","created":"2021-01-08 17:09:13.000000000","updated":"2021-01-13 12:17:45.000000000","submitted":"2021-01-13 12:15:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769965-osa/service_variables","meta_rev_id":"65da2253d82102cd1930da024aa28b1dfed5f4ee","_number":769965,"virtual_id_number":769965,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 12:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-01-13 10:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-13 08:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-01-13 10:02:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 19:14:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 08:41:19.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-01-13 10:02:58.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"157046b9a3c9cf67ad4bd0333620cea29b411d4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-08 17:09:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b578d44abd6007035ea9c5264d1919ceae7b2d94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 19:14:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4ac0bf339b124147b484be05c2d85c62 : FAILURE in 1h 56m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2594e6415b7244d191322dcc7beb3d50 : SUCCESS in 1h 39m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5896933c35d4dd5869fc83e7bca8293 : SUCCESS in 1h 35m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aea28a05a6f6466a81091b6ba629c258 : SUCCESS in 1h 27m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f86b624ecef544c7be8a0526d8507b10 : FAILURE in 48m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/113d1a2b7c014e66998d227606148998 : FAILURE in 1h 03m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8467b0448ee847dbaa59c5c9dfa80f1d : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"63e047648da9403c108d205c83780a3baff75246","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-11 16:15:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5c8b1bb6d0b6299d9b05e09e979f06de5ced2ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 18:09:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/deede8a1f9364f5d86244ef2feee2ac2 : SUCCESS in 1h 40m 20s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3c7683af800b4e85855d2236f943978d : SUCCESS in 1h 47m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42a3e06ac6f84bd2bd6b120fb486ae32 : SUCCESS in 1h 37m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb4abf650ace40c88d440b281c92d876 : SUCCESS in 1h 22m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/70b02bd5584a4bf78b6d2dd08d2ddab5 : FAILURE in 1h 02m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/464727cf1ab2463685195c744f29406f : FAILURE in 49m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a1506235474f5ebb5441528b3922e7 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"037374e005631e3d1d5570623937a2dc9a6c831f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-13 08:41:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3117d019c9826594c1ba46e75bde89acacf1f993","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-13 10:02:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff3542f2f0ac07414a3eb2b29050728b6ba72db2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 10:03:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d1e08893354f2dad6254c3e67f3fd713c1e306d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 12:15:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c4f30309ce344730a8d4573ec3231fcd : SUCCESS in 1h 40m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e3195c1ef2a14f0e8b333be2c0ffc0b8 : SUCCESS in 2h 02m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ddb49a155b540b2a6e0e43f07716506 : SUCCESS in 1h 50m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bc209a026a44941a03f503cd664946e : SUCCESS in 1h 39m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23515541a5f54415869247776464004b : SUCCESS in 7m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a204d2b92e2f488f9314fbe0255735b3 : FAILURE in 1h 02m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25c7ea7f56793fb4d876efde0c50f3afe24bc748","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 12:15:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"65da2253d82102cd1930da024aa28b1dfed5f4ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 12:17:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/405cf9513db74fe7909f413e8309b61d : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79f05091bdd48a3658efaf4b473e55511d64673a","revisions":{"79f05091bdd48a3658efaf4b473e55511d64673a":{"kind":"REWORK","_number":1,"created":"2021-01-08 17:09:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/65/769965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/65/769965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/65/769965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/65/769965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/65/769965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/65/769965/1"}}},"commit":{"parents":[{"commit":"a44292615808c4b42938df0475d63bcd272fe1af","subject":"Merge \"defaults: start kuryr service before docker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/a44292615808c4b42938df0475d63bcd272fe1af"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-01-08 17:09:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-01-08 17:09:03.000000000","tz":120},"subject":"Use global service variables","message":"Use global service variables\n\nInstead of overriding each service separatelly it might make\nsense for deployers to define some higher level variable that\nwill be used first or fallback to default variable.\n\nChange-Id: Iced1c0986f448b2a19b1f81586e50a039c23f566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/79f05091bdd48a3658efaf4b473e55511d64673a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/79f05091bdd48a3658efaf4b473e55511d64673a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
