)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~835907","triplet_id":"openstack%2Fopenstack-ansible-os_zun~master~Ia5ae1d9bbb6ad28fec9b0e7006630861fab5da19","project":"openstack/openstack-ansible-os_zun","branch":"master","topic":"workaround-setuptools-autodiscovery","attention_set":{},"removed_from_attention_set":{"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2022-03-31 15:11:43.000000000","reason":"Change was abandoned"},"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2022-03-30 21:14:10.000000000","reason":"Neil Hanlon replied on the change"}},"hashtags":[],"change_id":"Ia5ae1d9bbb6ad28fec9b0e7006630861fab5da19","subject":"Disable setuptools auto discovery","status":"ABANDONED","created":"2022-03-30 20:22:07.000000000","updated":"2022-03-31 15:11:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac03bf7c9cafa09ce96da305b112eeeefe9ef365","_number":835907,"virtual_id_number":835907,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-30 20:43:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":2,"date":"2022-03-31 13:26:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2022-03-30 21:14:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-30 20:22:07.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2022-03-30 20:43:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 13:26:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"fa887d19601a9f4e66e2521727ab988b475de3af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-30 20:22:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a975a7dabe4ed887fca1cbc8b65f4901277dd2cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 20:43:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbcdc806288242138c429f045e44a4a5 : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/128d1923edcf4e3487f24e0683ef4d37 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"bb4bbe49347dd1ad5442c6dfd507a1539a021a7c","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-03-30 21:14:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e369109b804600e096fb0184287879673d193fbc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-31 12:10:59.000000000","message":"Topic set to workaround-setuptools-autodiscovery","accounts_in_message":[],"_revision_number":1},{"id":"7b043ad37bffd0a135ccc3a42a40fd911131d4ed","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-31 12:51:44.000000000","message":"Abandoned\n\ncleanup instead","accounts_in_message":[],"_revision_number":1},{"id":"b972e5f24c23c64639c5abf4acc16e4d2e57cb2d","tag":"autogenerated:gerrit:restore","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-31 13:26:46.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"547100bd1adde80bdcefb7e8845bbbf20368ec12","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-31 13:26:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61f41204488f293356033f210ae45ca4097c3111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 14:02:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d3da381864e4879a71068f342532ca0 : SUCCESS in 21m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d7a356a6c9400ba76211551465cb16 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ac03bf7c9cafa09ce96da305b112eeeefe9ef365","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-31 15:11:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6766a2f39f7835e1ccb7d2030a32bb372dd54d9","revisions":{"e6766a2f39f7835e1ccb7d2030a32bb372dd54d9":{"kind":"REWORK","_number":1,"created":"2022-03-30 20:22:07.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/07/835907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/07/835907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/07/835907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/07/835907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/07/835907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/07/835907/1"}}},"commit":{"parents":[{"commit":"57fa739b905f2a3c8eb6d84e29f7f0adf7cafbcd","subject":"Merge \"Update Zun api-paste\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/57fa739b905f2a3c8eb6d84e29f7f0adf7cafbcd"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-03-30 20:22:02.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-03-30 20:22:07.000000000","tz":-240},"subject":"Disable setuptools auto discovery","message":"Disable setuptools auto discovery\n\nJobs are broken due to a release of setuptools 61.0 introducing\nnon-backwards compatible breaking changes [1]\n\nSetting py_modules to an empty string here will disable autodiscovery\nand should permit jobs to run as before.\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nSigned-off-by: Neil Hanlon \u003cneil@shrug.pw\u003e\nChange-Id: Ia5ae1d9bbb6ad28fec9b0e7006630861fab5da19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/e6766a2f39f7835e1ccb7d2030a32bb372dd54d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/e6766a2f39f7835e1ccb7d2030a32bb372dd54d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
