)]}'
{"id":"openstack%2Fopenstack-ansible-os_zun~867495","triplet_id":"openstack%2Fopenstack-ansible-os_zun~stable%2Fzed~I519f6b568a41d0679e7bacef10c5958a37c963fb","project":"openstack/openstack-ansible-os_zun","branch":"stable/zed","topic":"create-zed","hashtags":[],"change_id":"I519f6b568a41d0679e7bacef10c5958a37c963fb","subject":"Update .gitreview for stable/zed","status":"MERGED","created":"2022-12-13 13:22:15.000000000","updated":"2022-12-13 15:10:05.000000000","submitted":"2022-12-13 15:08:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867495-create-zed","meta_rev_id":"264852ae755a18aaa82d1881b2a92a7c21f34967","_number":867495,"virtual_id_number":867495,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-13 15:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-12-13 14:56:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-12-13 14:57:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2022-12-13 14:58:10.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-13 13:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-13 14:56:24.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"0f42b89837ab535d89b3ef3d55cb9186b6cc0f54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-12-13 13:22:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"345cca29b3cea9ab34d70bd8e5be4ca5845d7956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 13:32:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f031173807d544be8c052c196706e7f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9baf628aa96413a92f612899539dec3 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"80247d3e585897aedd62ca61a2dc059d33c31e3a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-13 14:56:24.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3667b422f06da944a8256dddcf7fc091d8ab689","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-13 14:57:57.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b8defda743a01e962b45634984dd34418a0028ec","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-13 14:58:10.000000000","message":"Patch Set 1: -Backport-Candidate","accounts_in_message":[],"_revision_number":1},{"id":"e2aff633e5f1853322c9b451a2968dbb9f84e322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 14:58:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd99060bbace700a8085b303a534ff4096dff8c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 15:08:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/679d9db10afd4c4686166ff4da5ce0ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26acdd4e79ca42d98f4bacb9c60eb518 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"97f89fb067e9a6b59008d5901d44a210c3f1ef32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 15:08:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"264852ae755a18aaa82d1881b2a92a7c21f34967","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 15:10:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/289e9aa742864f33a2abd85ae87353aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca6778d834e745ef89332916215b5d4e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66832ba3b37c18383371f0e9e5f9b0438edf7b8e","revisions":{"66832ba3b37c18383371f0e9e5f9b0438edf7b8e":{"kind":"REWORK","_number":1,"created":"2022-12-13 13:22:15.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/95/867495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_zun","ref":"refs/changes/95/867495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/95/867495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/95/867495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/95/867495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_zun refs/changes/95/867495/1"}}},"commit":{"parents":[{"commit":"53ad680494586eabaf14715eef7b2f0f062f5489","subject":"Move etcd installation to playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/53ad680494586eabaf14715eef7b2f0f062f5489"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-12-13 13:22:13.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-12-13 13:22:13.000000000","tz":0},"subject":"Update .gitreview for stable/zed","message":"Update .gitreview for stable/zed\n\nChange-Id: I519f6b568a41d0679e7bacef10c5958a37c963fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/66832ba3b37c18383371f0e9e5f9b0438edf7b8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_zun/commit/66832ba3b37c18383371f0e9e5f9b0438edf7b8e"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
