)]}'
{"id":"openstack%2Fopenstack-ansible-plugins~1004417","triplet_id":"openstack%2Fopenstack-ansible-plugins~master~Ibc93c629f9ecb0ba5d003f99f461b996b53d05c7","project":"openstack/openstack-ansible-plugins","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-09-14 17:58:34.000000000","reason":"Change was submitted"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-09-14 17:58:34.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-09-14 17:58:34.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-08 13:01:29.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-14 17:58:34.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-09-14 15:17:16.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ibc93c629f9ecb0ba5d003f99f461b996b53d05c7","subject":"Fix image upload with local filename","status":"MERGED","created":"2026-09-07 12:46:59.000000000","updated":"2026-09-14 17:59:31.000000000","submitted":"2026-09-14 17:58:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004417","meta_rev_id":"cdd2762be9de20aa858b8f8333939304b0bb29e5","_number":1004417,"virtual_id_number":1004417,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 17:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-09-08 13:01:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-09-14 15:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-09-14 15:17:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-08 13:02:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 13:07:03.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-09-07 13:07:03.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-09-07 17:23:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 13:02:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-09-08 13:02:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-09-08 13:02:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"462f15af8cd1e55a86f7d4b8102d1f9a00be7a80","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 12:46:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5733c213d04c36f6405571d96d25208ab8290c9d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 12:50:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"aa63a3d1501e3f34f3cb8a8c3732f32e530e7f69","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 13:04:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b1223ba0ffbed85c66cdd4e53a95982fa15f1565","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 13:07:03.000000000","message":"Patch Set 3: Backport-Candidate+1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"92c7f99bd148457c73819dfae816569633a84394","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-07 13:07:12.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d500131c425260c409d89ca2197ca8987ea70381","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-07 13:21:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0794f593f15f6b05fb9303e8d597b726b3b6f8d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-07 13:27:07.000000000","message":"Patch Set 4: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"bfcd53cc0d2feeb16bb4a2fffd6d815ac0f29fac","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-07 13:27:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b72e14f02892015d59ab9044b277fbc04c63f0f0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-07 13:32:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9ac822f4d34b588fc1d31fa91061cd093f5b37a6","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-07 13:56:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ff61065d2ba8ae71615ac3d1d5edcfe79b141563","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-07 14:10:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f6eaeeffefb4349647e50f511ff0e7dc2eef6b97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-07 15:30:48.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"634f3ccbc7ebea80f0c0d43c9bc382012d58a666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 17:23:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd03c077744246e5be64dab7243f266a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2083c14b1224486482a9cd498ec2703e : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30c7dbda5a4748bf949548488a41d4ef : SUCCESS in 2m 11s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/9e4974b3faec456c8a3ada9edf936756 : SUCCESS in 9m 09s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b5e95096ccf4f5a8be9e8f8ddaf5d15 : SUCCESS in 9m 57s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/51ef015972934a93822d2d3dd9b7d1c9 : SUCCESS in 5m 48s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/22f3e055feff4165a96b119e6bf4d80a : SUCCESS in 11m 06s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eea723f69187421c87e52e4efb100a4e : FAILURE in 5m 51s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/297292e7cc934c1fa54cdf704eadf88b : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/6d0ae4435f5044faa2e033ebe808e445 : SUCCESS in 52m 11s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b60729fa12044166921c489209339e14 : SUCCESS in 1h 43m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bbc54758dee24a69918d781ddf6ba9d2 : SUCCESS in 56m 25s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5c7e1eb653d142aab15cccb63742c97c : SUCCESS in 1h 19m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10bcf369af144a2192d93158f96be27a : SUCCESS in 58m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e81aee0c3a8c4e5a8fa9f10ecd4de33d : SUCCESS in 1h 06m 16s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24176bbbe24848ae9b6968040fef07e3 : SUCCESS in 1h 29m 40s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47104fc8540b4e22ad1f0d9468e130e6 : SUCCESS in 1h 41m 39s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/281decc9e624408fb698e58049c27712 : SUCCESS in 48m 57s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/59022c47aedd454cb3de31e35859bb7f : SUCCESS in 1h 02m 39s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/84dede23a7c546a1bcd5d6745f16b51b : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":5},{"id":"874a6985a51a4ee33af6fe4cd8f1b0bb9255c772","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-08 13:01:29.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1d18482a6303e34424579f09ffccd1510995159a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 13:02:57.000000000","message":"Patch Set 5: Backport-Candidate+1","accounts_in_message":[],"_revision_number":5},{"id":"800cd87093f43d3d4b6a9435ac73fd04dcc42c5f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-09-14 15:17:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a6cc06c0abff652c8a7e4dd5f70440334f94c22b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 15:21:27.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3592171b7eba22ccf043d4847ba5af1fa6f1102b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:58:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf433dafffac4aed91469b7fd4284d95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d36d67425c4f51911fbac6214a8b39 : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4135c6ef6a14a0996b46e9cd2bbd282 : SUCCESS in 3m 27s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/1d60f2c321ed487c884aa2bfca20be8a : SUCCESS in 12m 14s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5aae647669da432e93cbf37fd4682588 : SUCCESS in 10m 14s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/0cf0d89dc330485c97b443ee81219ed0 : SUCCESS in 11m 51s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/449a1bce4f6c443b8de56997bc04d4d3 : SUCCESS in 19m 51s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f089b2ef0d834b90b8d6852d1ba41124 : SUCCESS in 2h 23m 08s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a16212e7171b46c9bab561e4849768e7 : SUCCESS in 1h 06m 49s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/06ee782647014dbd8c0d3b9090ac4495 : SUCCESS in 1h 06m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1014464ff53d4bdcb62cb710a462f98f : SUCCESS in 1h 17m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9ce1546263c54c65a74127357108bf22 : SUCCESS in 1h 40m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e65cc5926614ca98a7e8d199b751ca8 : SUCCESS in 1h 07m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/05e065abdf29459390d41f052d939b94 : SUCCESS in 1h 48m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9a1089d1ae8410d9070c3d9ab4996fb : SUCCESS in 1h 53m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcde0614bdff470db02758b87863033d : SUCCESS in 1h 12m 25s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1597767b3bd34592b61649a0c847decc : SUCCESS in 57m 09s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5511a2b084224ef68019c072ef903f17 : SUCCESS in 43m 54s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/44b6e2f655f24429bb0a62da4069a5f1 : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":5},{"id":"4e6b2157bf03d677d3b65de53fbdece86f9ebfa6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:58:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"cdd2762be9de20aa858b8f8333939304b0bb29e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:59:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fda1afb3d45c40cd92e4428cf4ad8936\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c8a8debea14edaa110bc2ed78ed585 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5dafafe25fa9d339255b3c75e6348c09806c539f","revisions":{"b5acde40ec46d7c052399f943c808e44dcae2760":{"kind":"REWORK","_number":1,"created":"2026-09-07 12:46:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/17/1004417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/17/1004417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/1"}}},"commit":{"parents":[{"commit":"40370729165630af4c1d46a8e0b3c0a61f05558c","subject":"Merge \"Fix example for defining openstack aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/40370729165630af4c1d46a8e0b3c0a61f05558c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 12:46:59.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 12:46:59.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ibc93c629f9ecb0ba5d003f99f461b996b53d05c7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/b5acde40ec46d7c052399f943c808e44dcae2760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/b5acde40ec46d7c052399f943c808e44dcae2760"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"a541a90b3ee9d839a237be2a8c44af7b4547fda8":{"kind":"REWORK","_number":2,"created":"2026-09-07 12:50:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/17/1004417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/17/1004417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/2"}}},"commit":{"parents":[{"commit":"40370729165630af4c1d46a8e0b3c0a61f05558c","subject":"Merge \"Fix example for defining openstack aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/40370729165630af4c1d46a8e0b3c0a61f05558c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 12:46:59.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 12:50:50.000000000","tz":0},"subject":"Fix image upload with local filename","message":"Fix image upload with local filename\n\nChange-Id: Ibc93c629f9ecb0ba5d003f99f461b996b53d05c7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/a541a90b3ee9d839a237be2a8c44af7b4547fda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/a541a90b3ee9d839a237be2a8c44af7b4547fda8"}]},"branch":"refs/heads/master"},"96eca103f38f5a4875e53b65f72499562f3797f8":{"kind":"REWORK","_number":3,"created":"2026-09-07 13:04:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/17/1004417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/17/1004417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/3"}}},"commit":{"parents":[{"commit":"40370729165630af4c1d46a8e0b3c0a61f05558c","subject":"Merge \"Fix example for defining openstack aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/40370729165630af4c1d46a8e0b3c0a61f05558c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 12:46:59.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 13:04:08.000000000","tz":0},"subject":"Fix image upload with local filename","message":"Fix image upload with local filename\n\nChange-Id: Ibc93c629f9ecb0ba5d003f99f461b996b53d05c7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/96eca103f38f5a4875e53b65f72499562f3797f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/96eca103f38f5a4875e53b65f72499562f3797f8"}]},"branch":"refs/heads/master"},"291e78a8c362d37db84750a94d9555a34a2984c8":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-09-07 13:27:07.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/17/1004417/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/17/1004417/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/4"}}},"commit":{"parents":[{"commit":"40370729165630af4c1d46a8e0b3c0a61f05558c","subject":"Merge \"Fix example for defining openstack aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/40370729165630af4c1d46a8e0b3c0a61f05558c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 12:46:59.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-07 13:27:07.000000000","tz":0},"subject":"Fix image upload with local filename","message":"Fix image upload with local filename\n\nAllow local image uploading without requiring defining url with\ntemporary directory creation skipping.\n\nChange-Id: Ibc93c629f9ecb0ba5d003f99f461b996b53d05c7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/291e78a8c362d37db84750a94d9555a34a2984c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/291e78a8c362d37db84750a94d9555a34a2984c8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5dafafe25fa9d339255b3c75e6348c09806c539f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-09-07 15:30:48.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/17/1004417/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/17/1004417/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/17/1004417/5"}}},"commit":{"parents":[{"commit":"40370729165630af4c1d46a8e0b3c0a61f05558c","subject":"Merge \"Fix example for defining openstack aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/40370729165630af4c1d46a8e0b3c0a61f05558c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 12:46:59.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-07 15:30:48.000000000","tz":0},"subject":"Fix image upload with local filename","message":"Fix image upload with local filename\n\nFix image.url resolution error when uploading images with a local\nfilename, and skip cleanup when no temporary directory was created.\n\nChange-Id: Ibc93c629f9ecb0ba5d003f99f461b996b53d05c7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/5dafafe25fa9d339255b3c75e6348c09806c539f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/5dafafe25fa9d339255b3c75e6348c09806c539f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
