)]}'
{"id":"openstack%2Fopenstack-ansible-plugins~1005088","triplet_id":"openstack%2Fopenstack-ansible-plugins~master~Id0349659debf19e1704c154bc195051bc24cacba","project":"openstack/openstack-ansible-plugins","branch":"master","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-21 23:41:15.000000000","reason":"A robot voted negatively on a label"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-09-14 15:18:44.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Id0349659debf19e1704c154bc195051bc24cacba","subject":"Replace deprecated LXB","status":"NEW","created":"2026-09-10 16:45:21.000000000","updated":"2026-09-21 23:41:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af972e03298f917184aca2b57b2e949ae1319572","_number":1005088,"virtual_id_number":1005088,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-18 14:53:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":2,"date":"2026-09-18 09:38:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-14 15:18:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-09-18 09:38:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 18:58:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-11 12:39:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-09-14 15:18:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-14 15:18:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-09-18 09:38:08.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"702ea01db716135315dc9a1fc5c56c950ffc747a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-10 16:45:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0c0e7e9a2d2101a75f3576608616a97bf2b123d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 18:58:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a09de9a577646b4a15365e140bfd5fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e0d424b8294572b16ca29475975cd7 : SUCCESS in 2m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c13d300709b648eba1e4782a4b0eb659 : SUCCESS in 3m 15s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ce62bc5d3db74f9f81d6a67a911c5420 : SUCCESS in 7m 58s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aaf1fa8187f14dd7bfc295b4c6e742f2 : SUCCESS in 10m 01s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/0ec8be38fd124cfd9f4d3e0ec2553999 : SUCCESS in 11m 25s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/073b1318aebb448a899a86f219c28b89 : SUCCESS in 10m 43s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f1d53a1b66e54afa8bd40ee49e1ae491 : SUCCESS in 10m 20s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f14d95eaf0f4c13ae606a4e1d9c281d : SUCCESS in 2h 04m 42s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9da6b7be50a04d23af1b6a464e1425a6 : SUCCESS in 57m 48s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a4e6dad42b8445dcb8b5ea6a0ee12333 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6a68fe02df0944d99bbf10edb462aef8 : SUCCESS in 1h 02m 30s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dc9ea9f136a3497996bc3b46f7968c3b : FAILURE in 1h 20m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bd41289970a40749bced27f3d882a8b : SUCCESS in 1h 37m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7c851b2fd7ad42a2b222efeba83c8606 : SUCCESS in 1h 10m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22dde049977f43e1912bddf3ec2c5567 : SUCCESS in 54m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25a3bba5d0db436592a3aa5e550cba45 : SUCCESS in 53m 16s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a851906af4d49529bebcb3998991358 : SUCCESS in 49m 13s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6cd22c92ab1a4bc19153295eed4187e3 : SUCCESS in 41m 51s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6bad787685ca4032bb4470f211721c41 : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":1},{"id":"02d3830238e5a8261dbe789bd6ce1d8ea843613e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-11 11:15:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"537dbcc1257ecd9c0702dc721ab537576d3abacc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-11 12:39:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22c5abb5431ab6033b05cacf89d45f10c3a4b5cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 14:35:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/91ae44bb1bca4dd09edccef5383dca00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d9c80d93c204a0bb488666f32a31458 : SUCCESS in 2m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/903cf8f6f2c6442184be19fb9b1f2455 : SUCCESS in 3m 51s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/1db7051377e84706874ace7e0d300641 : SUCCESS in 10m 17s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63ede54a493f466e9ac08b7983b42e9e : SUCCESS in 22m 05s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/73d85d0f0b01403989ec2aec2f3f0be7 : TIMED_OUT in 31m 18s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/914c115762f94c58a1bd79a81c3982b7 : SUCCESS in 12m 58s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ee96679633f0483bb8275c4f00ac78b6 : SUCCESS in 9m 23s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8577b4153d449869e819434e1e8777c : SUCCESS in 2h 06m 05s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1975dc0fa6344dd8b7807127193a545d : SUCCESS in 55m 42s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c1f8d5befe364450bb25deffec77917e : SUCCESS in 1h 12m 28s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7bb59425a82a4bfea62d1cdf83f167f5 : SUCCESS in 1h 47m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6077d48a25184b7ca3cdfb1e996519fe : SUCCESS in 1h 24m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0eb700c9f7d049609d2823c499403b27 : SUCCESS in 1h 36m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/05f036445355460da2e06c07889a7ddb : TIMED_OUT in 3h 06m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb1a61e231af49618a6a217e31e0e9f0 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/335ac08d4517400c8b255403a8e562c2 : SUCCESS in 1h 39m 22s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/092d71cf4d244ed5924269e8cfcd43cf : TIMED_OUT in 3h 05m 03s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eac7cc41199e4407825b54b8f1ae9146 : SUCCESS in 46m 17s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/7259b3a48728444a9271946d853bd793 : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e70ccff88be368b7107a0a4973a532b6689324aa","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-12 13:54:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f47113b294a1ec83b03187ccc76b0f9cf3b04f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 16:05:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bbf7d675d314844ad8d90a0681097bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1d8b04cff849c78c9e14a8cd694875 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf00933fb6224d6e88e6101f3b21944a : SUCCESS in 3m 14s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ca111d46e7a448ac958fd3c8d8e46028 : SUCCESS in 9m 23s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59cb92bf5e3f480aa5b99c2db6c69b16 : SUCCESS in 9m 16s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/29c309fb7cd24055a9762c8134290255 : SUCCESS in 6m 37s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a0aead17878047769c4182f5097941f5 : SUCCESS in 12m 14s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d5510561f109456199c65e9063d5e44e : FAILURE in 5m 42s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c43243b4506f4679866407718dd625b6 : SUCCESS in 2h 02m 00s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1a5db2d118664e0795cc1a88cb348bac : SUCCESS in 1h 10m 52s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/4337c9eb32ce4904b7af8aca4a9eb4af : SUCCESS in 1h 10m 22s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aff9b93defe64358a70078b4ab2eabcf : SUCCESS in 57m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b2102d04367b4518ba8f10e2726c2226 : SUCCESS in 56m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21111817b6724713a98ba0507298c8f5 : SUCCESS in 1h 28m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ae03cdbdb132431bb60a5dc993494f31 : SUCCESS in 49m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5ae77037e204f3e80fc31f7ff95ce63 : SUCCESS in 1h 29m 47s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e25c488c05de4282a9f17d6a8aaaaa2b : SUCCESS in 1h 35m 43s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71d05aa3707d421d94526434a210d78e : SUCCESS in 47m 42s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a0ef6294b37648a79e1557c33c0cfb57 : SUCCESS in 33m 53s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3a8086a820ff4790b5a066aee855a47e : SUCCESS in 47m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d38d681260b2db6d444f188b3ce81c368cc381e8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-14 15:18:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4409b310d30843090d4e10e53b7c78b82bdc0308","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-09-18 09:38:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb0cb943b8801e45e99bfe25d4e4e39fbdf1dea7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 09:43:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54d0bf525ec72de20b1392057d6d8c0ec0778f27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 12:19:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/326339db19494276a4ff0f5bebd663bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15c092b9ca9489caf1f517cab49bb6e : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57cf45e414694a42a8a5fa32c7d723f9 : SUCCESS in 5m 01s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/0922ea5183eb479788df74b1f01653f0 : SUCCESS in 9m 32s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b28b9266ee4424c8a9ebbfffa040630 : SUCCESS in 14m 05s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/cd3938b173b84945813f7723cd18c205 : SUCCESS in 10m 45s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/394ce7f301e04d74b61224c1a9a719c0 : SUCCESS in 10m 49s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1f2ec465e0644169bb5ef840a1d77c1 : SUCCESS in 2h 30m 18s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f137fd983faa4ab5917ce8abd29f7cef : SUCCESS in 1h 53m 17s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b01c250855c74b949e160e56e405ba5c : SUCCESS in 1h 41m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/be133323cab8461b85e81f1166cafda7 : SUCCESS in 1h 37m 16s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/53506b5fcd5043eea4f58a358f07be64 : SUCCESS in 1h 41m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5689bb3520e6431b874beb3c486c9d61 : SUCCESS in 1h 45m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d1aa32ae4d714493818b66a3ee2e98c6 : SUCCESS in 1h 48m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ea46ac0f9ed4b109f761064d2ea0c67 : SUCCESS in 1h 54m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c6855b4090449c5b3a3c123aea1e033 : SUCCESS in 1h 35m 22s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c4b4cf0f82e4334a9cd814010f5876b : FAILURE in 58m 36s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ab886e4be0a64e2b8fb264dabc3e1fa4 : SUCCESS in 1h 06m 25s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/56b555ce001244e797677dd27836bdea : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3f376df2aa3d531d2b483e10037b9ddadf000f5f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 12:51:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9de0b36527bff77f7fb7cddfe0b42ab601177e9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 14:53:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c9ca8c9880f4160b6d55a88726b89c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be287c47a778448d9740338b8348a74d : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ceaa0f02f734a4fbba0e433a0d4d616 : SUCCESS in 2m 46s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/9b9f6bfee6034fc39b3f68bbe2b20d7f : SUCCESS in 12m 29s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d206ac906bf34b1186e9dd7ab13abaed : SUCCESS in 6m 27s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b5e723d417114a53b96fc08d1802b792 : SUCCESS in 10m 35s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/afe715c4c1374ed28c5b852eee1b5e60 : FAILURE in 6m 31s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/33e712d7a256431c8daf7f4dafbbd50d : SUCCESS in 9m 54s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc227495da1d4d73aeeb52e16092cb0d : SUCCESS in 1h 47m 58s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/67a339dfafd8495f8516fdeefc2b5a04 : SUCCESS in 57m 46s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/bbc26ea201b942cda274ef19594e99d6 : FAILURE in 1h 19m 49s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2e0163f8d872417297691578372801c1 : SUCCESS in 1h 54m 53s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a6d4489d86e441e996cd527154888214 : SUCCESS in 1h 55m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/baf9d625eade4d599978285ec335f9c1 : SUCCESS in 1h 42m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2108c7703ce74226b086206fe85139d1 : SUCCESS in 1h 40m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bac5a035edc42a1945dc10969eccd5e : SUCCESS in 1h 38m 13s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec6094b5bf9840539585e30b7bcc12a3 : SUCCESS in 1h 42m 08s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/471c3c006b4143a9b6868c856ac1dceb : SUCCESS in 29m 29s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fa5a473db2d6490fb11b464df51ee3bd : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b5b3da18c4654cf9a21d5c097dff86f8 : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d39557078695387d31b6fd23a67911e8e7b3ad49","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-21 21:16:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af972e03298f917184aca2b57b2e949ae1319572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 23:41:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c5f43a11a2a4d559de047fd5cde6efb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd82f47471f34654b18bfed92984ae44 : SUCCESS in 2m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dec489828014944a61d1e387a4808dc : SUCCESS in 2m 18s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/e9560a51293a479bbdb22e896a18683a : SUCCESS in 10m 11s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af79d86fef384745978b23fde8f97092 : SUCCESS in 6m 22s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/00ca09a26bb1498eaafafbb647a126a5 : SUCCESS in 11m 25s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/fd68a4440c964e4e811032c54e8198d4 : FAILURE in 8m 22s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/abfa14e2dc5147e48e66b341acd8a451 : FAILURE in 6m 05s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29357b39eefc42b4bf807e1ac2722786 : SUCCESS in 2h 11m 10s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3558f69ab9684ac7a6c88cc8c4657161 : SUCCESS in 1h 28m 56s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/aa654af2061946abab40fccb4a9caa6d : SUCCESS in 1h 36m 51s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bbefdf15cf294065907295d4f0b71028 : SUCCESS in 1h 13m 57s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b35ed3cb39d44ca9af49b4a0c5c5740d : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06eae80a9cec4712a031a31a6f91effa : SUCCESS in 1h 37m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/3fa22fd38c3d41e29a21c09b9ea7f2f9 : SUCCESS in 52m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2e0d34eebaf436d92bb9fd3a41c27e9 : SUCCESS in 1h 46m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b7deb2eaa384421825bac3856d65c40 : SUCCESS in 1h 08m 27s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/965af08c48174c368ab3621a128e2bab : SUCCESS in 32m 05s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/26d1e7ca68ce4276be02dbeea3d00a0e : SUCCESS in 37m 50s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/502959233c014ed19f3e3bb70f902628 : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c42c1f282925601aa0be8ce1c718a0b25af3f03","revisions":{"6c42c1f282925601aa0be8ce1c718a0b25af3f03":{"kind":"REWORK","_number":1,"created":"2026-09-10 16:45:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/1005088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/88/1005088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/88/1005088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/88/1005088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/88/1005088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/88/1005088/1"}}},"commit":{"parents":[{"commit":"40370729165630af4c1d46a8e0b3c0a61f05558c","subject":"Merge \"Fix example for defining openstack aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/40370729165630af4c1d46a8e0b3c0a61f05558c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-10 16:45:13.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-10 16:45:13.000000000","tz":120},"subject":"Replace deprecated LXB","message":"Replace deprecated LXB\n\nDespite linuxbridge_agent has been deprecated and removed,\nour playbooks and modules still were reffering to it throughout codebase\n\nIt is now replaced with appropriate to the context components.\n\nChange-Id: Id0349659debf19e1704c154bc195051bc24cacba\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/6c42c1f282925601aa0be8ce1c718a0b25af3f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/6c42c1f282925601aa0be8ce1c718a0b25af3f03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
