)]}'
{"id":"openstack%2Fopenstack-ansible-plugins~1005352","triplet_id":"openstack%2Fopenstack-ansible-plugins~master~Ica0b2dfc6169c430f32ee0f8d710b3e6bc709eda","project":"openstack/openstack-ansible-plugins","branch":"master","hashtags":[],"change_id":"Ica0b2dfc6169c430f32ee0f8d710b3e6bc709eda","subject":"docs: Add descriptive titles to action/filter plugin docs","status":"NEW","created":"2026-09-11 21:32:32.000000000","updated":"2026-09-11 23:54:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8554f69a57dd7325b9055328c1ddbef8dc8affcf","_number":1005352,"virtual_id_number":1005352,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 23:54:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 23:54:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed0242d263dc3955b276405a7af56247b71b5341","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-11 21:32:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f3d9c18141ebf9b296436fcba2988c751a3abae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-11 22:00:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"400f9dc41975bdb8f915463a4e53c72cdc5a3e1e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-11 22:00:54.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"07c82f2f9b02b495d80dec74b87a2089aad1a69e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-11 22:05:32.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"8554f69a57dd7325b9055328c1ddbef8dc8affcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 23:54:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f215b27f29fc47ed9cc7ebc9ecb2acf1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ee52eb1711f4f7aa868ac19eba6cea6 : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4319cc21c9a24820b275bf18369066e2 : SUCCESS in 5m 00s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/428f7e4e3155444abb4700c84fdc9129 : SUCCESS in 9m 36s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f894924f27214b128683efc0894bde5f : SUCCESS in 11m 05s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e019cc6415e948be9376636ce8ce1a99 : SUCCESS in 10m 54s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f66cf2e4814f41b6bb9b752595a6b5b3 : SUCCESS in 12m 50s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/53e093db740d469abc16bed4281f85a4 : SUCCESS in 9m 03s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/294c50694917439891dfa45b7a1bd3a0 : SUCCESS in 1h 32m 12s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/7d0693eb69174ad897f28cd815d6f498 : SUCCESS in 1h 34m 56s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9510c2210ac847d8bc100b959c49f309 : SUCCESS in 1h 13m 04s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e63a13751c054bf1823277fce673675c : SUCCESS in 58m 14s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1785283222bd4780b1c6e84b05b29562 : SUCCESS in 1h 19m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30d26b4dace8450ab0aab315a1b446ce : SUCCESS in 1h 34m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/cc037eabd80b43efaded3832e2b25eed : SUCCESS in 1h 35m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea910ae15dd848a4a9a3bf8795976c08 : SUCCESS in 1h 30m 15s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9799adf0931484e8233dca20d3eed0f : SUCCESS in 1h 11m 01s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aad6800894d0469f84a93ee8340246da : SUCCESS in 50m 41s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dad261f92ace4dd594301b74527db194 : SUCCESS in 36m 30s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/429cdd13e56e4dee9c2139a204feeb35 : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b303278e5afe52ba3089478247b7bfee1d165f33","revisions":{"c1c1eed601ce8ba7c092e3f0899c02e60dc6d2e3":{"kind":"REWORK","_number":1,"created":"2026-09-11 21:32:32.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/52/1005352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/52/1005352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/1"}}},"commit":{"parents":[{"commit":"40370729165630af4c1d46a8e0b3c0a61f05558c","subject":"Merge \"Fix example for defining openstack aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/40370729165630af4c1d46a8e0b3c0a61f05558c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-11 21:32:32.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-11 21:32:32.000000000","tz":0},"subject":"wip:","message":"wip:\n\nChange-Id: Ica0b2dfc6169c430f32ee0f8d710b3e6bc709eda\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/c1c1eed601ce8ba7c092e3f0899c02e60dc6d2e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/c1c1eed601ce8ba7c092e3f0899c02e60dc6d2e3"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"16342ee64d02cc2cf8eda51bff2e3d35aa1de91d":{"kind":"REWORK","_number":2,"created":"2026-09-11 22:00:52.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/52/1005352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/52/1005352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/2"}}},"commit":{"parents":[{"commit":"40370729165630af4c1d46a8e0b3c0a61f05558c","subject":"Merge \"Fix example for defining openstack aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/40370729165630af4c1d46a8e0b3c0a61f05558c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-11 21:32:32.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-11 22:00:40.000000000","tz":0},"subject":"docs: Add descriptive titles to action/filter plugin docs","message":"docs: Add descriptive titles to action/filter plugin docs\n\nRename action/filter plugin subsection titles to include\na short description of what each one does, instead of just\nthe plugin name.\n\nChange-Id: Ica0b2dfc6169c430f32ee0f8d710b3e6bc709eda\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/16342ee64d02cc2cf8eda51bff2e3d35aa1de91d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/16342ee64d02cc2cf8eda51bff2e3d35aa1de91d"}]},"branch":"refs/heads/master"},"b303278e5afe52ba3089478247b7bfee1d165f33":{"kind":"REWORK","_number":3,"created":"2026-09-11 22:05:32.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/52/1005352/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/52/1005352/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/52/1005352/3"}}},"commit":{"parents":[{"commit":"40370729165630af4c1d46a8e0b3c0a61f05558c","subject":"Merge \"Fix example for defining openstack aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/40370729165630af4c1d46a8e0b3c0a61f05558c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-11 21:32:32.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-11 22:05:26.000000000","tz":0},"subject":"docs: Add descriptive titles to action/filter plugin docs","message":"docs: Add descriptive titles to action/filter plugin docs\n\nRename action/filter plugin subsection titles to include\na short description of what each one does, instead of just\nthe plugin name.\n\nChange-Id: Ica0b2dfc6169c430f32ee0f8d710b3e6bc709eda\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/b303278e5afe52ba3089478247b7bfee1d165f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/b303278e5afe52ba3089478247b7bfee1d165f33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
