)]}'
{"id":"openstack%2Fopenstack-ansible-plugins~1005594","triplet_id":"openstack%2Fopenstack-ansible-plugins~stable%2F2026.1~Ibc93c629f9ecb0ba5d003f99f461b996b53d05c7","project":"openstack/openstack-ansible-plugins","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-15 08:17:10.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-15 18:15:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc93c629f9ecb0ba5d003f99f461b996b53d05c7","subject":"Fix image upload with local filename","status":"MERGED","created":"2026-09-14 15:24:41.000000000","updated":"2026-09-15 18:16:59.000000000","submitted":"2026-09-15 18:15:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005594","cherry_pick_of_change":1004417,"cherry_pick_of_patch_set":5,"meta_rev_id":"987792002fecdf1b5dee32f78e17eca6cda5da8a","_number":1005594,"virtual_id_number":1005594,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 18:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-09-15 08:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-14 15:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-09-15 08:17:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 15:25:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-09-14 17:45:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-15 11:17:45.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"}],"messages":[{"id":"efaed3e2dad760ee6d9cf06c4427c91ad587857a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-14 15:24:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"12811087522635755a9b241c48dcd438fb0d4956","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-14 15:25:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dda7f5272d7b456fbb4f5c007297ccddd3225f64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:45:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a8cde61d22147feaf3889ebb346fe3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77883e88bb724d8cb50f36ae9b75a1be : SUCCESS in 2m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a49bbab59bc04eafb7b8f9b1199b3ec6 : SUCCESS in 3m 10s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/46245fe075fa42fa96a62c194966508f : SUCCESS in 12m 17s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/090720a1c381471fa8e71b13febeeb92 : SUCCESS in 13m 11s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fe79b962e4e4a348f1371ec73af2e9c : SUCCESS in 11m 02s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/725515412ea549f49137b216064e01be : SUCCESS in 12m 50s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/bfdf713f81db4271b6645c4f808d8bab : SUCCESS in 13m 04s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f966ebc32efb41ab9e287af16f7987b1 : FAILURE in 10m 07s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1876e452d5f4faba6c87e2f7b8c4283 : SUCCESS in 2h 14m 52s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1da149594e064582ab35262d1d1a317a : SUCCESS in 1h 34m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/17762a8cb17c465f86f32c60ac8e92a2 : SUCCESS in 1h 40m 01s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/7318602c41c04087aa3bfc640f6c917f : SUCCESS in 1h 40m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/54a65d10d4234d69bef9781b356adb24 : SUCCESS in 1h 46m 45s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/60e216d7479b460b8675483785df7ddd : SUCCESS in 1h 51m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93e6d168829f48c59a105bb034c7087e : SUCCESS in 1h 51m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/252adaffaf13425590a5e51ada8c5a25 : SUCCESS in 1h 27m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13ae61ab075249e183e118114882cd62 : SUCCESS in 1h 37m 47s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7204d86a1a141009f9772f037723201 : SUCCESS in 1h 42m 14s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f6f3b162c62467b92cc1869902aa1d2 : SUCCESS in 52m 33s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/63f52bb0e9fa49f583bcd975c74b1504 : SUCCESS in 38m 07s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/52bc6dd03a184277b1e9c499aa7de39e : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":1},{"id":"05d7a3a1e2e5dae2c1b0865a4b790b5f4bf161e3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-15 08:17:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60b71b8015ff2164146ca8f0179591431b82248f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 08:21:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98030edd4e77fab39bec1efd9c4127f63e2c95ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 10:26:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18d85f1d9e9f4188ae0f2ba8a99be652\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/076da0e9d1a24ed592699dd98cde1c9c : SUCCESS in 2m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c7caf01d9d340af94b9e0a568aea5df : SUCCESS in 2m 05s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/f7931fadfae349e2a649f96fe3f21380 : SUCCESS in 12m 18s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/5d1595e4cc1e4e11adf4a287fc8e951b : SUCCESS in 12m 10s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc436f37a1614c7290bab58cf2c3ca92 : SUCCESS in 8m 39s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b96800bd7e194c44bf8e75f81418331d : SUCCESS in 12m 19s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/5a273241301640bd8cd355b893405279 : FAILURE in 8m 41s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fcb6846b2514f3aae2b705a8a9a3d67 : SUCCESS in 2h 03m 58s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/84831f5125e640c9907e256316da174c : SUCCESS in 1h 42m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/999ce0d953724198b9340bc08aaa1e9e : SUCCESS in 1h 42m 40s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1053c4e7aec945ac9dc7733b3ba70f76 : SUCCESS in 1h 41m 43s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c8d60b6205154037a69fdc5e279510e4 : SUCCESS in 1h 49m 54s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6d3e5f22b3f04ace949055357b248bbf : SUCCESS in 1h 53m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2b727d1d3644685a82e74b2a73f2000 : SUCCESS in 1h 40m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/8b057bc00b76414db741f8b5535b8cfc : SUCCESS in 59m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f2eb0981f31479db1f9e24d605f5edf : SUCCESS in 1h 41m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/335c33b3dd1e4d2badf2023775f0225c : SUCCESS in 1h 42m 42s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c51475ed88574a89a9f16c280a9046bf : SUCCESS in 31m 57s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c87985a8b4a8460fa17196831338c791 : SUCCESS in 1h 00m 59s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6217e48caa894ecbbe328aae8af42b71 : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":1},{"id":"101e95bef29d03145ea010fba9b54ad9ade87b70","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-15 11:17:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46181e671852c31d3b80c917806596a68d3da58b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 13:32:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f99cdc0d6c394f67ac4561763d15ae69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfeefe4b50ef4797a2838469700a788c : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b14f2116d8f04362a6ef8533809feffd : SUCCESS in 3m 15s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/b198d53d80664d3eb2e3eb949a9f88d8 : SUCCESS in 11m 38s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/f68e30f0976a41b6b1e7d08c0a38dc84 : SUCCESS in 7m 08s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4669c2d5095f4e22b63193b69b739bbf : SUCCESS in 8m 41s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5f98e87a10ee4930816ef8b432db22ab : SUCCESS in 12m 58s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/734e8f09e612429aa9e9c1af2019a1cc : SUCCESS in 13m 07s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/84fda3f0a24144bda6b3ec30b2aa2ed2 : SUCCESS in 13m 17s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cf440f6fb8d455fb04ae8847efefdaf : SUCCESS in 2h 06m 07s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/0e09cba41edb4a7b95450672683cc251 : SUCCESS in 1h 21m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/aa779c31ce30405292e99d9a26f298af : SUCCESS in 55m 36s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/15102597e9374548891bf9efabc12ff4 : FAILURE in 31m 34s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/edd62693658740bba1262685ed6a6961 : SUCCESS in 1h 19m 19s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/64a9b19a84da46ebac7a4c9cfe7909c1 : SUCCESS in 1h 04m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdbbb986b261448c97ad35fa06d556ee : SUCCESS in 1h 23m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/159dcb8f579b4150af2895ebdd1bf2e1 : SUCCESS in 1h 35m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16cff02738954aa19e09b8e560e27adf : SUCCESS in 54m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bee0f25cf9f43aa96df366737bc53f5 : SUCCESS in 1h 40m 10s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9442ecd57ceb47968e6645c4fde8d4c2 : SUCCESS in 48m 27s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c8642b7b8d19474fa61d2df00ef38315 : SUCCESS in 43m 18s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b44389259d6a47feb760ec51e4174603 : SUCCESS in 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"fc8b0d81c42efcbfaf0781dcd0ec1e02da21afb0","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-15 14:02:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e03accf2afb82b6aa899f85ec44598483c7ae7cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 16:12:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8c3df9d66c04b37aef197ed2f42fb98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e2ecbde51ca494d941dbb6f903fc389 : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af5be2a781a94eceb067b27530bf24a7 : SUCCESS in 3m 12s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/580e88407b6a4ad4bc71e8bb1dfb3d3c : SUCCESS in 17m 15s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/43118badfab340c48e4239b102cc6eb2 : SUCCESS in 10m 53s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a64abdc8abc4103844fe1d33eaa529d : SUCCESS in 14m 00s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e2c5d08ee2334384bd5bf557e6f16d62 : SUCCESS in 14m 07s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/64433be0eec24beeb204e135d9a8a94b : SUCCESS in 15m 56s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3f86cccb54f543fe8c0d4437b8f9b02d : SUCCESS in 13m 25s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d01c7bd40e6f462b9fdc85e461e57d86 : SUCCESS in 2h 05m 15s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/811d5396607b471988c2f6f35c0bcf6e : SUCCESS in 1h 21m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2214f8f8eddc483094aa3b11250cf595 : SUCCESS in 59m 05s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/fe5583e60db44e2aaa7c6fab6a038643 : SUCCESS in 1h 36m 36s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0b62ec33efa4487f8eeed598ea6a0f21 : SUCCESS in 1h 02m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/69fe6b50f27349aca069b186d5c5ba15 : SUCCESS in 1h 05m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/780b7c31fbb0430195949bf234b6f4b5 : SUCCESS in 1h 04m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b9ce35ca2f49493ab4ce469bf36e9bff : SUCCESS in 1h 36m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/507f84c738ab46c29e6d07d4ba8b79ed : SUCCESS in 1h 44m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db9a9e917ff74e48bcfb9832ed7113e2 : SUCCESS in 1h 35m 11s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78066978abcb4d639e0348735b081f3d : SUCCESS in 48m 18s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/919c85cd8dd84dd7912cc91a57b91820 : SUCCESS in 40m 32s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1e7782e4ba2545b98610c667b1d14d26 : SUCCESS in 32m 54s","accounts_in_message":[],"_revision_number":1},{"id":"93672a8a0581d562f8937241a7af36629ada271c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 16:16:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc3e6abf7f333e094ca61df1068cc91cc31f3fa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:15:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43a7a42042724f2990f1fb2abb6e52a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8578267d7bb447bca86c94b7b3cb275f : SUCCESS in 2m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf42a9e749e747e297867c09b61cb04e : SUCCESS in 3m 08s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/88204beb98fc407fb3dca438856ee1ea : SUCCESS in 10m 15s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/a79e58182c964802a219535858f971e8 : SUCCESS in 10m 59s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b065675fe37d4012a65dd22e17f2b1d0 : SUCCESS in 11m 09s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/72c9a05b0ece4a4db76457a06eccce3b : SUCCESS in 12m 25s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/68e292e1dcf64331b0fc261a544a09b0 : SUCCESS in 14m 30s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72ed11e331f04a65a7a0dab9ac75d387 : SUCCESS in 1h 21m 19s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/76dd3129546b4d11b72f8c9c6cb63586 : SUCCESS in 59m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/aa69670066a243ac8ccbc4cee70e8909 : SUCCESS in 1h 30m 22s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2071c306964646e889d0e6177a62c93b : SUCCESS in 1h 35m 35s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8ff6c0e2227a4e7fbca35b4e61ba19f0 : SUCCESS in 1h 00m 54s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/06b19bf8520349a68fdcee331aeb1ea6 : SUCCESS in 1h 08m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27419bd5b47e46779c91f17b049da45a : SUCCESS in 1h 04m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f2edb50c39ad45ffa0075d048349ced7 : SUCCESS in 1h 03m 02s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b00b0bef9da4a219e59e2a58cdfdf2f : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a326446cdc747a9a5e0805a165e6bcf : SUCCESS in 1h 37m 09s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32704cae427240f792be2f15462c4ff3 : SUCCESS in 54m 21s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1e0b522e3796436ba187a1c344c78ef6 : SUCCESS in 41m 37s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0f1064a7955d45fc860fff2904b6c7e4 : SUCCESS in 37m 07s","accounts_in_message":[],"_revision_number":1},{"id":"aef39050802c9e7e5ca4ce5888aafa323d33c775","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:15:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"987792002fecdf1b5dee32f78e17eca6cda5da8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:16:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/317cc24a33d34c369cbcb16c754bde4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37dd7850627443b28cecd46d3491aa58 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7e96032ef4721bbb8e13bcba26f1c465bb9761f","revisions":{"a7e96032ef4721bbb8e13bcba26f1c465bb9761f":{"kind":"REWORK","_number":1,"created":"2026-09-14 15:24:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/94/1005594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/94/1005594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/94/1005594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/94/1005594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/94/1005594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/94/1005594/1"}}},"commit":{"parents":[{"commit":"3d0cbebe281d57337e8073896373a1699d207f88","subject":"Merge \"Remove override for PrivateDevices\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/3d0cbebe281d57337e8073896373a1699d207f88"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 12:46:59.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-14 15:24:41.000000000","tz":0},"subject":"Fix image upload with local filename","message":"Fix image upload with local filename\n\nFix image.url resolution error when uploading images with a local\nfilename, and skip cleanup when no temporary directory was created.\n\nChange-Id: Ibc93c629f9ecb0ba5d003f99f461b996b53d05c7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/a7e96032ef4721bbb8e13bcba26f1c465bb9761f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/a7e96032ef4721bbb8e13bcba26f1c465bb9761f"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"40370729165630af4c1d46a8e0b3c0a61f05558c","ours":"3d0cbebe281d57337e8073896373a1699d207f88","theirs":"5dafafe25fa9d339255b3c75e6348c09806c539f","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
