)]}'
{"id":"openstack%2Fopenstack-ansible-plugins~1006360","triplet_id":"openstack%2Fopenstack-ansible-plugins~master~I48b962e7a78c8df590e062d36770ffd979a5d7f6","project":"openstack/openstack-ansible-plugins","branch":"master","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-19 09:03:32.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I48b962e7a78c8df590e062d36770ffd979a5d7f6","subject":"Use dynamic include for adriacloud.kubernetes roles","status":"NEW","created":"2026-09-19 08:45:17.000000000","updated":"2026-09-19 09:03:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"496b1a4da0da5fd886fb50815a1a6bf2fe654647","_number":1006360,"virtual_id_number":1006360,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-19 09:03:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-19 09:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8179c166dfc4f45280a1be2070709c362bcac85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-19 08:45:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"496b1a4da0da5fd886fb50815a1a6bf2fe654647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 09:03:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14a596e71de34750becf53daef19bd05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47cefbbce130473b995401251b8a420d : RETRY_LIMIT\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65261d1b9cc04ca09ea782ec7693b26d : RETRY_LIMIT\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d9c92151737b485cbfdf3eef675b9227 : RETRY_LIMIT\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2779c46eaf3e42ca84e06542c31d547c : RETRY_LIMIT\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/fe9e34eb80de461c80de899c49f017ac : RETRY_LIMIT\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/446414f341274b7c8de50bb80fe86882 : RETRY_LIMIT\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7c6e0986ada4498cb2bd0bdb02853e19 : RETRY_LIMIT (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/376720bc4f3641e8bd5021f62ea3767c : RETRY_LIMIT\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/708f4788d1c94f1eab86bdc9afa743ba : RETRY_LIMIT\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/cecefbc8704c48fda59d236fc01b88f0 : RETRY_LIMIT\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ba890fe0667b45d9936df89f44b6a8c9 : RETRY_LIMIT\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0cf27477b3114c42a0ed30d62242bf3c : RETRY_LIMIT\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2b335b1058f4db9960f3a3df07dc4bd : RETRY_LIMIT\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4940e661da7f41508fd619354b2edd78 : RETRY_LIMIT\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d37944bfb6984b5c838fb3c560a7e940 : RETRY_LIMIT\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3a9379f54904f23aebc378c86ba4739 : RETRY_LIMIT\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84ba042a00e643959b52fb83be9ea913 : RETRY_LIMIT\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d4b544fe77647c3b5fb8973f7093773 : RETRY_LIMIT\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d2a70f89f9f4480a8bcd7e56e1213fb3 : RETRY_LIMIT","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9139e5c3d2a22835e406bbcad4b83b85e2026529","revisions":{"9139e5c3d2a22835e406bbcad4b83b85e2026529":{"kind":"REWORK","_number":1,"created":"2026-09-19 08:45:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/60/1006360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/60/1006360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/60/1006360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/60/1006360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/60/1006360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/60/1006360/1"}}},"commit":{"parents":[{"commit":"0d055eb17cc04234f1008d0c5bc6d84f3690022b","subject":"Merge \"Document openstack_user variables for openstack_resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/0d055eb17cc04234f1008d0c5bc6d84f3690022b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-19 08:45:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-19 08:45:05.000000000","tz":120},"subject":"Use dynamic include for adriacloud.kubernetes roles","message":"Use dynamic include for adriacloud.kubernetes roles\n\nSome orgs may have stricter then others process for fetching external\nrepos and establishing trusted sources.\n\nOur k8s playbook heavily relies on 3rd party collections for k8s\ndeployment.\nAt the same time, we could easily bypass fetching these external\ndependencies while not failing setup_infrastructure if we\u0027d use dynamic\nincludes for the related to the collection content.\n\nChange-Id: I48b962e7a78c8df590e062d36770ffd979a5d7f6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/9139e5c3d2a22835e406bbcad4b83b85e2026529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/9139e5c3d2a22835e406bbcad4b83b85e2026529"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
