)]}'
{"id":"openstack%2Fopenstack-ansible-plugins~1006922","triplet_id":"openstack%2Fopenstack-ansible-plugins~master~I267778598dfd563e0be46d66ca2c363497df7db1","project":"openstack/openstack-ansible-plugins","branch":"master","topic":"osa/molecule","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-24 18:12:37.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I267778598dfd563e0be46d66ca2c363497df7db1","subject":"Address ansible-galaxy fetching issues.","status":"NEW","created":"2026-09-23 10:01:52.000000000","updated":"2026-09-24 20:32:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f2453ba9580fe4bbc809dd1d940a2ec190ae8c10","_number":1006922,"virtual_id_number":1006922,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-24 20:32:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 13:14:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4c44aeeafee300065d430c159e0adc247ccfc22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 10:01:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b8bff85cf56b050d0337e2a36b73ac3cbbc9097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:14:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a7024b5962a4167bf7574b87f3ad4f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dfaa13fdb55471590651e99bbb80f2c : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/288dca564a874c67b806abcf14a7c40d : SUCCESS in 3m 30s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/23262832be1345559c0af13791c67e7a : FAILURE in 8m 38s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1ca182100e44e7bb373fdbf42b2651a : FAILURE in 6m 21s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a7748cdd8c674066a9d9e8a257bd90a3 : FAILURE in 7m 26s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/ebd6bef92cfa463d9a519626ffecfd71 : FAILURE in 11m 21s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f7c8dcb3bae410285cdf4c7ff0ee8fc : FAILURE in 5m 57s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/278c0dd212824baca11a9fb50810efb3 : SUCCESS in 2h 19m 05s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c5930c4e858344c4bee5a83453172fd4 : SUCCESS in 1h 31m 40s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8a054db14ce443d5b785c36125f25714 : SUCCESS in 1h 43m 23s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d5c669d817546c894544dc65a135fc9 : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1d6ee717ef9d46d08f5e5183d4c3296e : SUCCESS in 1h 43m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9685516b935446dba14d33dfdc95ef52 : SUCCESS in 1h 39m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/be545aacfcc34edaa425b325d6b84f16 : SUCCESS in 1h 42m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b9fe0813f174ba396b884d0cbc39462 : SUCCESS in 1h 57m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3076876d85c04c689cec76c7fccb6643 : SUCCESS in 1h 58m 24s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f1402640cd24db38d78739d4c2e131a : SUCCESS in 57m 13s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5103e147a1014604a597f4ded534dc53 : SUCCESS in 54m 18s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/52f5b83d070f4b95b9fb94da9f7d10e9 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":1},{"id":"47b3e81cbec4eb91b10fd247413cc5cc3e54bcfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-24 09:39:30.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a81535f93001ee1061f232a216a71d513f0cd717","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 12:04:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca1c8e316a5446aba31bf00f1b4c0a7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a88c0b48a0a4d738d28ae1cf5e84ce4 : SUCCESS in 2m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/446f5a2e0ad24bd9bad457b3ae99f534 : SUCCESS in 3m 54s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/6a8185d6ea1745779360ec219746d3de : SUCCESS in 9m 12s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e51d2b9bd0224228b9c8583965a83627 : SUCCESS in 9m 48s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b718f4677a284c2693ed0e288e1f23ee : SUCCESS in 12m 42s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2830f712474e40a78b2648dd4a957e49 : FAILURE in 6m 10s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/896046a34e8841ae93e4f0e4cf730021 : FAILURE in 5m 54s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6848edd954f14ae88855ae9f6ad682d5 : SUCCESS in 2h 13m 58s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d1e4b1f917e84e30876435a304f568f0 : SUCCESS in 1h 11m 08s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/174cbbce9e154e2b9b65454f5cfde75e : SUCCESS in 1h 42m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/894dcb3983224f79a078004b64268830 : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9858269638e046bba46cfea0e290adc5 : SUCCESS in 1h 57m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/501fb2ab77b746f7915ed455581e1d88 : SUCCESS in 1h 51m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/965a4f7533854972ac36dfffa84ffad6 : SUCCESS in 1h 26m 51s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75879f6f05b74a6ca08a115003821b00 : SUCCESS in 2h 00m 51s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3075dfb625e844548eab05996505c6c8 : SUCCESS in 2h 06m 33s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0cea274c31343ad9852793b9759ddf2 : SUCCESS in 54m 07s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/19988c5d202344b6be8e1592607b16c6 : SUCCESS in 59m 21s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9b750985df4f4a58945fa375c5411431 : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":2},{"id":"6aef2349b8a4cf474eb54448b13f329effb509da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-24 14:49:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8eb815713f16c4b51d8c9ee4303128ab09af168b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:21:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f63d77888bbc42f98ec111bec154b6a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5502567dd3d544c7befee73dd745c695 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3af1680da6f34ea196d68d968113a6e4 : SUCCESS in 3m 09s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/02d50471208640dea32a280094e37634 : SUCCESS in 6m 00s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbd481931cdb46db97d287c2dd4c7b9e : SUCCESS in 7m 21s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/713589c94e1648efb4a1f8847c9cfea0 : FAILURE in 9m 44s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/ca0d8416a9fc4c02a21070548f179808 : SUCCESS in 10m 32s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7d1a76bce2284e7895a07e7fcc95b015 : FAILURE in 5m 53s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/357a3e8ba1cb49ae8ec865313275f184 : SUCCESS in 2h 19m 29s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/7abb3698f457451aa39d9ca5d2c90151 : SUCCESS in 1h 50m 54s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c88880019eb2430ca421db82fdaab75b : SUCCESS in 1h 29m 19s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d57fb5a65c0f45d8aaacdeeb331f8782 : SUCCESS in 1h 35m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3e79d1fa532144cca2a408e97efad820 : SUCCESS in 1h 08m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce4025a189274b06b37e690313c99218 : SUCCESS in 1h 37m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d0d07909e4b443e2917ce6d77f1c06f0 : SUCCESS in 1h 07m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f79f4ca97cba469899a78542f6d4af1f : SUCCESS in 1h 05m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5edc2fd5f5d24b49a4c0944fad59ed5a : SUCCESS in 1h 04m 18s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f1fd10ee2954e01bbd2981a6efde181 : SUCCESS in 1h 01m 12s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c98ba7f52a4f49a79cba00c6e0c02fc6 : FAILURE in 53m 57s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/84c72e964592465c95c107511a662bd5 : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":3},{"id":"45143abd3102bd8b6e8f799f8fd3d731b485d5f2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-24 18:12:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2453ba9580fe4bbc809dd1d940a2ec190ae8c10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 20:32:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9519609997c54b1c829761eda0969f34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b75b61a91ff465e99cc4d2c4e39424a : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b7164ad6a2d4e1a911292b5b405992b : SUCCESS in 4m 40s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/55e98d7ca3324ee2930c9d233d15581e : SUCCESS in 9m 07s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02cbf705116b4dbf9d5666262193e322 : SUCCESS in 8m 50s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/0901c34dd31842df8294538ffb8f3566 : SUCCESS in 10m 50s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/0c1a5d3416aa467ebf8547f46139170c : SUCCESS in 7m 56s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c67e0568b3be468d9726eb6b81eddce1 : FAILURE in 5m 33s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9317d19f4489445daa58b66a4a3fc5f9 : SUCCESS in 2h 08m 21s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/457577c9e2a54c058612cdc7c1b6a274 : SUCCESS in 57m 54s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/77f245a4ea714dbeb2a83531ff917060 : SUCCESS in 1h 35m 22s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/46f385ed89df45e29463bbfd535d7d63 : SUCCESS in 1h 51m 34s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e8002f37ba924e7eb335a58f07ceb0c7 : SUCCESS in 1h 30m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3edc1d9bf13487983641bfc48a3e47e : SUCCESS in 1h 43m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/dd0c6cad215344b3a29bfced34eaa081 : SUCCESS in 1h 48m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/304b7aefca62415d9f53b99871c3370e : SUCCESS in 1h 23m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7639053f36534577a90ea0783716cb07 : SUCCESS in 1h 22m 02s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd5fc0e5a526416ea0d2d8ef17b55f74 : SUCCESS in 50m 58s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ffe015458aec4d68b7e393d51a1da308 : SUCCESS in 33m 05s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/56fc82fdaf9c4d128b9c2385904933e2 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7607091e3aaf0f468f24603e4012809e31be8334","revisions":{"c19c94c82b184238587461c8bbfd03d5f7e32fc9":{"kind":"REWORK","_number":1,"created":"2026-09-23 10:01:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/1006922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/22/1006922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/1"}}},"commit":{"parents":[{"commit":"0d055eb17cc04234f1008d0c5bc6d84f3690022b","subject":"Merge \"Document openstack_user variables for openstack_resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/0d055eb17cc04234f1008d0c5bc6d84f3690022b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-23 10:01:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-23 10:01:35.000000000","tz":120},"subject":"Address ansible-galaxy fetching issues.","message":"Address ansible-galaxy fetching issues.\n\nIntermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\ntests/requirements.yml, which is parsed by default with\nansible-compat, and it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10\nStream.\n\nChange-Id: I267778598dfd563e0be46d66ca2c363497df7db1\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/c19c94c82b184238587461c8bbfd03d5f7e32fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/c19c94c82b184238587461c8bbfd03d5f7e32fc9"}]},"branch":"refs/heads/master"},"e67ab042e1d1e0b3dbd0fddf3add5ad7f0aff0d1":{"kind":"REWORK","_number":2,"created":"2026-09-24 09:39:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/1006922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/22/1006922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/2"}}},"commit":{"parents":[{"commit":"0d055eb17cc04234f1008d0c5bc6d84f3690022b","subject":"Merge \"Document openstack_user variables for openstack_resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/0d055eb17cc04234f1008d0c5bc6d84f3690022b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-23 10:01:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-24 09:39:25.000000000","tz":0},"subject":"Address ansible-galaxy fetching issues.","message":"Address ansible-galaxy fetching issues.\n\nIntermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\ntests/requirements.yml, which is parsed by default with\nansible-compat, and it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10\nStream.\n\nChange-Id: I267778598dfd563e0be46d66ca2c363497df7db1\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/e67ab042e1d1e0b3dbd0fddf3add5ad7f0aff0d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/e67ab042e1d1e0b3dbd0fddf3add5ad7f0aff0d1"}]},"branch":"refs/heads/master"},"7607091e3aaf0f468f24603e4012809e31be8334":{"kind":"REWORK","_number":3,"created":"2026-09-24 14:49:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/1006922/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/22/1006922/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/22/1006922/3"}}},"commit":{"parents":[{"commit":"0d055eb17cc04234f1008d0c5bc6d84f3690022b","subject":"Merge \"Document openstack_user variables for openstack_resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/0d055eb17cc04234f1008d0c5bc6d84f3690022b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-23 10:01:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-24 14:49:38.000000000","tz":120},"subject":"Address ansible-galaxy fetching issues.","message":"Address ansible-galaxy fetching issues.\n\nIntermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\ntests/requirements.yml, which is parsed by default with\nansible-compat, and it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10\nStream.\n\nChange-Id: I267778598dfd563e0be46d66ca2c363497df7db1\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/7607091e3aaf0f468f24603e4012809e31be8334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/7607091e3aaf0f468f24603e4012809e31be8334"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
