)]}'
{"id":"openstack%2Fopenstack-ansible-plugins~664356","triplet_id":"openstack%2Fopenstack-ansible-plugins~stable%2Frocky~I6dc82ff0ed4f0f51a3285ab5b7425f5c2762afca","project":"openstack/openstack-ansible-plugins","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I6dc82ff0ed4f0f51a3285ab5b7425f5c2762afca","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:45:36.000000000","updated":"2019-06-10 17:41:31.000000000","submitted":"2019-06-10 17:41:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664356-1560188491100-7fce86ad","meta_rev_id":"3c93bf56aec2d9c8544197f0433e7afa1fba74a1","_number":664356,"virtual_id_number":664356,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 17:41:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:18:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:18:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:18:12.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 17:41:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c1f3ccbaa948b8a4f327a79cb2be009d4004cf7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:45:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9185cd6389707668d2252d0c074eaea8a59e18a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:44:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/56/664356/1/check/build-openstack-sphinx-docs/fafa687/html/ : SUCCESS in 4m 51s\n- openstack-ansible-linters http://logs.openstack.org/56/664356/1/check/openstack-ansible-linters/af4bac8/ : SUCCESS in 6m 48s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/56/664356/1/check/openstack-ansible-functional-centos-7/fd233ab/ : SUCCESS in 20m 36s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/56/664356/1/check/openstack-ansible-functional-opensuse-423/d422c7f/ : FAILURE in 8m 43s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/56/664356/1/check/openstack-ansible-functional-ubuntu-bionic/81e1135/ : SUCCESS in 18m 23s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/56/664356/1/check/openstack-ansible-functional-ubuntu-xenial/9976512/ : SUCCESS in 24m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bc28df8227d6368f1622afd2e037ffc0cd52ab7d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:18:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e385028a44d0d9708646db35fcd3fcf3b46ddded","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:18:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce3db377467ef4ca42112691facaee8357ccdee6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:41:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/56/664356/1/gate/build-openstack-sphinx-docs/f24e873/html/ : SUCCESS in 3m 45s\n- openstack-ansible-linters http://logs.openstack.org/56/664356/1/gate/openstack-ansible-linters/11ff4df/ : SUCCESS in 5m 45s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/56/664356/1/gate/openstack-ansible-functional-centos-7/d0bdacc/ : SUCCESS in 17m 17s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/56/664356/1/gate/openstack-ansible-functional-opensuse-423/8714371/ : FAILURE in 9m 05s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/56/664356/1/gate/openstack-ansible-functional-ubuntu-bionic/855ffb9/ : SUCCESS in 16m 46s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/56/664356/1/gate/openstack-ansible-functional-ubuntu-xenial/5d4408a/ : SUCCESS in 15m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4b71bbaad24e71d7422de18d6896964d5db961bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:41:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f8a1552f5b6bc889e8013b5a40f4bc2c31191c5","revisions":{"3f8a1552f5b6bc889e8013b5a40f4bc2c31191c5":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:45:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/664356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/56/664356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/56/664356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/56/664356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/56/664356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/56/664356/1"}}},"commit":{"parents":[{"commit":"be7c455237e4264f524902c7283a0d59a65f38ab","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/be7c455237e4264f524902c7283a0d59a65f38ab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:35.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I6dc82ff0ed4f0f51a3285ab5b7425f5c2762afca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/3f8a1552f5b6bc889e8013b5a40f4bc2c31191c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/3f8a1552f5b6bc889e8013b5a40f4bc2c31191c5"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
