)]}'
{"id":"openstack%2Fopenstack-ansible-plugins~997173","triplet_id":"openstack%2Fopenstack-ansible-plugins~master~I9fc72042638ae271dd28a6fb3d18fbd594ea8a14","project":"openstack/openstack-ansible-plugins","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-27 16:00:15.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-08-27 13:04:36.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I9fc72042638ae271dd28a6fb3d18fbd594ea8a14","subject":"Re-enable CentOS 10 Stream jobs","status":"MERGED","created":"2026-07-14 07:52:33.000000000","updated":"2026-08-27 16:02:56.000000000","submitted":"2026-08-27 16:00:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997173","meta_rev_id":"9afdfb5beab0a16fda9512d24f4a1c66aaa8a206","_number":997173,"virtual_id_number":997173,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-27 16:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-26 12:04:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-08-27 13:04:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-08-27 13:04:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-14 10:19:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-14 17:55:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-08-26 12:04:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-08-26 12:04:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"dbe5fbc8803776f9d4a36b719da59be9fa9dca3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-14 07:52:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36928b246c18b6bd7418fbb1a79b4fbfa30c7a05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 10:19:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e184b60e356c4917b615a21a02a517f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d82b0e662304b2ab3d84b123c224c5f : SUCCESS in 2m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f55fe92ce4c84dee96c261f5f0694eaf : SUCCESS in 2m 22s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/6951d77bacd8464a94636fe5bd75de77 : SUCCESS in 9m 28s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2098c4763d4545f890b8ee5669b918f8 : SUCCESS in 9m 55s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4b1a9b578bc148a9a516ddccb2da2d25 : SUCCESS in 10m 41s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/256e5d62a57e4056b077f25f5b6faf6d : SUCCESS in 8m 34s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/65ae8090cdfd4d3cb2509a6ec3ddfb4b : FAILURE in 9m 34s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9d8211e53c944c99634a2523a92759e : SUCCESS in 2h 07m 20s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b2b13a86db834863a509bf1aa31cea2c : FAILURE in 23m 20s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/cf168824a5814ed890f5d0799d555af0 : SUCCESS in 1h 32m 59s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5882e56c8ad14e37a05a7d835045bf6f : SUCCESS in 1h 00m 10s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/07996299410641a2b1d8a55594b375a6 : SUCCESS in 1h 02m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de88fdd7642f4ec8a9cdabc261f36eba : SUCCESS in 1h 55m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a44883a4219f49e09dd99a324f1a6af3 : SUCCESS in 56m 20s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2c6d0a6979e4bb188eadbaa11ec8ff6 : SUCCESS in 1h 38m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbdb4cdc7646403a83d5626fd4ce0405 : SUCCESS in 2h 19m 04s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d9fde1bc2234ea49df692465f393207 : SUCCESS in 52m 04s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/431adcda88e64d3d9d31c85d4718c6e0 : SUCCESS in 47m 37s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ff8fa29df9e94aaa969be465489c42c4 : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":1},{"id":"10f5fc3fdc49523c6647bf0993e8b92c88dd699c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-14 12:35:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf7fe515c0859841ad175fb23e1bbb0a7d610676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 16:02:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bbd5cea96244fae8509e94de433d9f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131bb0d281d14f16b28fb516c61421f2 : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce90b9441c5f475db97302f6d13b8b75 : SUCCESS in 3m 30s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/03166529ba2a4b9aa0c42fe8ed158527 : SUCCESS in 7m 12s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7459a35e2cfe451fa4df80265c1604cb : FAILURE in 5m 08s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c2db34c5b4d743ac96ce53756fde338d : SUCCESS in 9m 33s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/8a18aaa81d5c483c969b4f7d78391bb5 : SUCCESS in 9m 04s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f95c2000bc024af8bad82ae2710157bf : FAILURE in 5m 35s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f117a8582dee48179bdcf5b99f03a7a9 : TIMED_OUT in 3h 08m 05s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ae25f67257284b4d9401e1593777f9aa : SUCCESS in 3h 18m 49s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1f6a138f808c4f19a5369e8a1bd59d1c : SUCCESS in 1h 35m 02s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/543bdc6eb3854c2fbcc0d854a90f4822 : SUCCESS in 58m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/71872d86bee74d6b9295540083f4936c : SUCCESS in 1h 47m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94e1d4083765444689fa910bdce9ceef : SUCCESS in 1h 40m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7e2e99c8a7804b87a49cc031c083a5d9 : SUCCESS in 2h 23m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba19f70963c64a37a85e8ba266aed4a6 : SUCCESS in 1h 35m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23a4a360de264bfbbc0bb1ede9790d23 : SUCCESS in 1h 38m 37s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f6b9bdd7ca54223a43236ebfff9f27f : SUCCESS in 1h 06m 05s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fbf38c74c10748129fb4546591343302 : SUCCESS in 48m 23s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/5a5d5247237d499eae0605096053d449 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8a203e143fd662814410dc767350a944f958f80d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-14 17:55:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0858f15f49ed50d13fd472bda38ac697c96e22e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 21:17:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dde770acb5144591a9a567994847b795\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5778edfda1bf4110916c68c5e604ae8a : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f94eaed237964b59b7b9105d5623ca73 : SUCCESS in 2m 45s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ffcaca1d5a05435caa5fe5a5fb2b082f : SUCCESS in 9m 44s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/464d0667be924fa8a9d13903db2e7906 : SUCCESS in 10m 00s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e7706a6e445d4b99ba0dc772692d5709 : SUCCESS in 6m 25s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/26cb0371ef5a48798042e8fe8fd2d65e : SUCCESS in 8m 27s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/89f44be55d6d49e98b9604e36aed7b0b : SUCCESS in 7m 57s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c424a94862404578aae03c0d57221a41 : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/6fb157669a3c4f2192e1d78654e73a13 : TIMED_OUT in 3h 11m 49s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/87f464117bdf4c809bc49836f5e85054 : SUCCESS in 1h 18m 21s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/04f3411051424cfaa7863ec01c0590a0 : SUCCESS in 1h 17m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/067ac5fe5bb1416a91c82448321bb8ae : SUCCESS in 1h 19m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b09b7015cf8b413393a1c10b6b569b0a : SUCCESS in 50m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/447c9356cdde4fbe92625489c1fd4444 : SUCCESS in 1h 35m 47s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94cb96c6291f461d8affb5567072f2f4 : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15d8ad41b4f342a28f1c8895b7c05835 : SUCCESS in 1h 37m 07s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96f67bafaa8f4bf685095ad8d43c684a : SUCCESS in 52m 36s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c61f1e3ed29b46f099b8d01d2b437430 : SUCCESS in 44m 35s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2df1be9e1181499691272c79a40ec98f : SUCCESS in 51m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4be31ab8c185619620f44e4a50c5396577ac9fac","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-14 23:28:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"468c8d6c876c71916ff2397782430a1c8d68cb7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 02:54:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95bf00aaa6a248518b9b22802bfef3a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b271f688ce41debf4cd8d4f1116cfe : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b6910858083451b83bccd879360228b : SUCCESS in 1m 54s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/0d8b47357c244ba3917e397091b49cd5 : FAILURE in 6m 57s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0aee6764b6c54674ab3a79910b06f68b : FAILURE in 6m 44s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/fcbe414f38f044f9a4622cfda50359a5 : FAILURE in 8m 49s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/c5f7c055617f4896b29101f7f164e7b2 : FAILURE in 5m 40s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fecc84b64c3b4aecb2abe671d6a9d324 : FAILURE in 6m 29s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf6c0becdda64175ae6ef72c380c24ed : SUCCESS in 2h 35m 42s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/5f3b0bc09cea4434aae9b3e21367dfd4 : TIMED_OUT in 3h 21m 07s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f4f1bec1c5cf4ab3a3c8b739582c046c : SUCCESS in 1h 33m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/493164de32164eabb4c55c6bd5d315b6 : SUCCESS in 1h 08m 09s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9449590382ea49cd85fde706b3592f4b : SUCCESS in 1h 01m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95dbbbcdffdc4acaa9b884e13448ad9a : SUCCESS in 1h 29m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6ec39705a9e64dbc9f6bd6312b16f964 : SUCCESS in 53m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4b5e9fb496b4668994418e3e1b696dc : SUCCESS in 1h 44m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a4e0df941dd42f8a3377106c52b3939 : SUCCESS in 1h 32m 18s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da7ebc3475884750901a15865f7a338a : SUCCESS in 53m 49s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/85cfe3af24fb45ac8ce51708ad1f72a0 : SUCCESS in 42m 58s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/62384e84b3ed4803a425f2d5e73ff325 : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5752a0f84089d1196a38142f902f666086e166d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 09:37:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_28619\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"_revision_number":2},{"id":"5e7883df5c8221d2c1bdc119ff157a42785b5c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 11:59:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5197b47fced4461a003923ea2360a99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db7435fa1d648f282c445ae070f7156 : SUCCESS in 2m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cb08e0cd08246b18b90bfc5fe3d0da9 : SUCCESS in 3m 04s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/decb202524f1483c8fd852f568caef8c : SUCCESS in 10m 06s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71ed6089b19640308b2fc38511aac06b : SUCCESS in 11m 19s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4d239db8ac934d3586ac822b692cc08d : SUCCESS in 6m 01s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/199edde0119b4f11ba85714d3ac32f5a : SUCCESS in 7m 26s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2131ab3e46ca49f4857d1e676ba2c480 : FAILURE in 6m 49s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30d3fe8f60af4c7f8fe3361490026c61 : SUCCESS in 2h 11m 58s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f9119fc76dba40b89c36faf7d00ef315 : SUCCESS in 1h 01m 22s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/5defe5d27d2543b28f5d66460d956dff : SUCCESS in 1h 35m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/10ea4202b3ea465e9da31f98574e1131 : SUCCESS in 1h 52m 17s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cd4808f7c0884c47b4bc67f7022b0751 : SUCCESS in 1h 07m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c64e2a2ece774998b771d4949c017946 : SUCCESS in 1h 42m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/3f2b5cb2a7d04ff6ac2d2554bb39205f : SUCCESS in 1h 12m 35s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03a5c55d3cc4477db57891751136d414 : SUCCESS in 1h 35m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f8be5fab2004f0db9cffb0246811d0b : SUCCESS in 1h 02m 41s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bce3d03577b4a909c6369e21d62af7c : SUCCESS in 52m 24s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2496fcf2d479431580dbba19ed47fba7 : SUCCESS in 38m 41s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b7c94561cde14c6f9b2ecc04e590d06d : SUCCESS in 42m 39s","accounts_in_message":[],"_revision_number":2},{"id":"05cb3eee3a88fae4c71289d1e59bd229e4740846","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 12:04:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cd1960677d500f6bc7913dbd937f7c96a156d022","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-08-27 13:04:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c33354eea95dc69f82b66e43abf34b0028f22fc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 13:10:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8682cad761d09522c445bb97e1358167fc891a69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 16:00:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75a176dc2f4748758b409a7dc2842211\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889e2dc768964eb29b025f85c351f523 : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bff0d6de92a04417b4ba60d4525c8e6b : SUCCESS in 3m 19s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/adc3aaee8fc94a6a81c93e4d9a7d1123 : SUCCESS in 9m 42s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbe834349a3549f998217f0d4729ccd2 : SUCCESS in 7m 06s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9b8371219eec4f4fab2977ba906eba04 : SUCCESS in 11m 42s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d9893fa527804112b5508ac3426f2fd6 : SUCCESS in 15m 20s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cba168f34b34f058eaf9c7817414423 : SUCCESS in 1h 11m 36s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/0f9649720178433d8d06e1fdb309d3c9 : SUCCESS in 1h 03m 21s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8d17d9101f5f42b6b6b01a386b0cbebf : SUCCESS in 57m 29s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3e213f55fef3447382b316dfe04644dd : SUCCESS in 1h 52m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fc5354c4163c4b15a467d08cb57945c9 : SUCCESS in 1h 59m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92683443590c46bf86130541ba03ea24 : SUCCESS in 57m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/eda84ac865d14e43a8587addf5a1dee3 : SUCCESS in 2h 30m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8307fadfd2b466b9676679f364b394a : SUCCESS in 1h 57m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68ddc4672c8a46a69a7e63712523c9f5 : SUCCESS in 2h 39m 32s\n- openstack-ansible-deploy-k8sHA_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/145bfd3415f442f18940a2d12b0dccde : SUCCESS in 2h 00m 40s\n- openstack-ansible-deploy-k8sHA_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9000b374e50a44bfb12c9f01990391b2 : SUCCESS in 38m 35s\n- openstack-ansible-deploy-k8sHA_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2f60f093a1d848e4aaceffa08d31476a : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":2},{"id":"a688c9d59f6220ace20254b8f8279d5a0bdcbdba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 16:00:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9afdfb5beab0a16fda9512d24f4a1c66aaa8a206","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 16:02:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc6b6b7e1bbd412f96f40748683cbffb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7cb3e2766b42f4b62d77857039e7b2 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82d1af3bf5b69726686089de887175121c300010","revisions":{"8116d4bbd7cf48126367a68045f1a481f88406cb":{"kind":"REWORK","_number":1,"created":"2026-07-14 07:52:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/73/997173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/73/997173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/73/997173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/73/997173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/73/997173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/73/997173/1"}}},"commit":{"parents":[{"commit":"62819f38cf90f544a2e7a1ea6a0f1d5985a65650","subject":"Replace repo with utility for hosts healthcheck","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/62819f38cf90f544a2e7a1ea6a0f1d5985a65650"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-14 07:52:26.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-14 07:52:26.000000000","tz":120},"subject":"Re-enable CentOS 10 Stream jobs","message":"Re-enable CentOS 10 Stream jobs\n\nChange-Id: I9fc72042638ae271dd28a6fb3d18fbd594ea8a14\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/8116d4bbd7cf48126367a68045f1a481f88406cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/8116d4bbd7cf48126367a68045f1a481f88406cb"}]},"branch":"refs/heads/master"},"82d1af3bf5b69726686089de887175121c300010":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-26 09:37:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/997173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-plugins","ref":"refs/changes/73/997173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/73/997173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/73/997173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/73/997173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-plugins refs/changes/73/997173/2"}}},"commit":{"parents":[{"commit":"3ea0ed99b0100733cf1146a9a03ba4fedf17baef","subject":"Merge \"Fix compatability with latest molecule-plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/3ea0ed99b0100733cf1146a9a03ba4fedf17baef"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-14 07:52:26.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-26 09:37:08.000000000","tz":0},"subject":"Re-enable CentOS 10 Stream jobs","message":"Re-enable CentOS 10 Stream jobs\n\nChange-Id: I9fc72042638ae271dd28a6fb3d18fbd594ea8a14\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/82d1af3bf5b69726686089de887175121c300010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-plugins/commit/82d1af3bf5b69726686089de887175121c300010"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"62819f38cf90f544a2e7a1ea6a0f1d5985a65650","ours":"8116d4bbd7cf48126367a68045f1a481f88406cb","theirs":"3ea0ed99b0100733cf1146a9a03ba4fedf17baef","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
