)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~432042","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~master~Ib308ed698b7c09ef3277fe74957f1ed91aa1b9da","project":"openstack/openstack-ansible-rabbitmq_server","branch":"master","topic":"openstack/requirements","hashtags":[],"change_id":"Ib308ed698b7c09ef3277fe74957f1ed91aa1b9da","subject":"Updated from global requirements","status":"MERGED","created":"2017-02-10 05:56:38.000000000","updated":"2017-02-10 08:11:26.000000000","submitted":"2017-02-10 08:11:26.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9c5398b75a5d2ab06b8f0d8a408771da0a7229db","_number":432042,"virtual_id_number":432042,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-02-10 08:11:26.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-02-10 08:07:13.000000000","_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":2,"date":"2017-02-10 06:12:12.000000000","_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-02-10 08:07:13.000000000","_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-02-10 06:12:12.000000000","updated_by":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"reviewer":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"state":"REVIEWER"},{"updated":"2017-02-10 08:07:13.000000000","updated_by":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"reviewer":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"state":"REVIEWER"},{"updated":"2017-02-10 08:11:26.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"01760ab5e3ed600c5815891a9eaad4e2f75e6cce","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-02-10 05:56:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1523ededb17e2ee3834957b980d107641431610","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-10 06:05:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial http://docs-draft.openstack.org/42/432042/1/check/gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial/3c1ee35//doc/build/html/ : SUCCESS in 1m 43s\n- gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial http://logs.openstack.org/42/432042/1/check/gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial/a8f223b/ : SUCCESS in 2m 30s\n- gate-openstack-ansible-rabbitmq_server-releasenotes http://docs-draft.openstack.org/42/432042/1/check/gate-openstack-ansible-rabbitmq_server-releasenotes/d530a08//releasenotes/build/html/ : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c3eafde45e050de572034a1cca74c21837d4e8ad","author":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"date":"2017-02-10 06:12:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3334d3ff5d7708e137955f495859b570695657c","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2017-02-10 08:07:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d68137871e858d004aba1feafd16343d7f138a2c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-10 08:07:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"1d31582a6f3afa8399173efe0a49caedc909728a","date":"2017-02-10 08:11:26.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"278e33390312cb542bbec0cda0f4aaf7f0336009","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-10 08:11:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial http://docs-draft.openstack.org/42/432042/1/gate/gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial/e34e790//doc/build/html/ : SUCCESS in 1m 15s\n- gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial http://logs.openstack.org/42/432042/1/gate/gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial/1e704b8/ : SUCCESS in 1m 03s\n- gate-openstack-ansible-rabbitmq_server-releasenotes http://docs-draft.openstack.org/42/432042/1/gate/gate-openstack-ansible-rabbitmq_server-releasenotes/8587d64//releasenotes/build/html/ : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da549dee7d23272a1463cd09a6fdf2670a8ed9b0","revisions":{"da549dee7d23272a1463cd09a6fdf2670a8ed9b0":{"kind":"REWORK","_number":1,"created":"2017-02-10 05:56:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/432042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/42/432042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/42/432042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/42/432042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/42/432042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/42/432042/1"}}},"commit":{"parents":[{"commit":"4f36ee4cae25180309c76e3039a90927b0573f5e","subject":"Update RabbitMQ to v3.6.6-1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/4f36ee4cae25180309c76e3039a90927b0573f5e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-02-10 05:56:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-02-10 05:56:37.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: Ib308ed698b7c09ef3277fe74957f1ed91aa1b9da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/da549dee7d23272a1463cd09a6fdf2670a8ed9b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/da549dee7d23272a1463cd09a6fdf2670a8ed9b0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4f36ee4cae25180309c76e3039a90927b0573f5e","is_merged_in_target_branch":true,"change_id":"I4847382d614f582944e8711ab7b5b37740f9b3ce","change_number":430604,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
