)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~441327","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~master~I7953a8eafb2d9595f7c3758cddeb3f513246bfc8","project":"openstack/openstack-ansible-rabbitmq_server","branch":"master","topic":"python2-pyOpenSSL","hashtags":[],"change_id":"I7953a8eafb2d9595f7c3758cddeb3f513246bfc8","subject":"Install python2-pyOpenSSL package on CentOS","status":"MERGED","created":"2017-03-03 18:50:05.000000000","updated":"2017-03-03 20:29:32.000000000","submitted":"2017-03-03 20:29:32.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4578c2e4d3dfa010ea7d32e65d48757f9741d2be","_number":441327,"virtual_id_number":441327,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-03-03 20:29:32.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-03-03 18:52:42.000000000","_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":2,"date":"2017-03-03 20:25:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":1,"date":"2017-03-03 20:25:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-03 18:52:42.000000000","updated_by":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"reviewer":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"state":"REVIEWER"},{"updated":"2017-03-03 20:25:57.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2017-03-03 20:29:32.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ba805420fa6aedac6a9574a20c8dd63cc3d38fb2","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2017-03-03 18:50:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e38d28110d35f64893caf8474bc8114bd714591e","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2017-03-03 18:52:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a3a1c4cb918e67836cc21b4508e0b8cf914ce26","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-03 18:59:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial http://docs-draft.openstack.org/27/441327/1/check/gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial/6af7a7c//doc/build/html/ : SUCCESS in 2m 28s\n- gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial http://logs.openstack.org/27/441327/1/check/gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial/0cf52a1/ : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1},{"id":"63f9b5ae90e85ceb59194f880039a3d0a2fa176d","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2017-03-03 20:25:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50d8b7ae91d07dbd6648c5feb8df15142161796b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-03 20:26:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"c198cb3ccb0a6321aaea27781da1c37468125468","date":"2017-03-03 20:29:32.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"2fb1412a2709bda0e8412d43328ba5721f27144d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-03 20:29:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial http://docs-draft.openstack.org/27/441327/1/gate/gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial/175a5fd//doc/build/html/ : SUCCESS in 2m 07s\n- gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial http://logs.openstack.org/27/441327/1/gate/gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial/e629f9f/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"93160e01ddbe370f64b2cceb25b888ac369d736e","revisions":{"93160e01ddbe370f64b2cceb25b888ac369d736e":{"kind":"REWORK","_number":1,"created":"2017-03-03 18:50:05.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/27/441327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/27/441327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/27/441327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/27/441327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/27/441327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/27/441327/1"}}},"commit":{"parents":[{"commit":"f02d890466859e2e14e6e030eee1b7620194a9c7","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/f02d890466859e2e14e6e030eee1b7620194a9c7"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2017-03-03 18:48:06.000000000","tz":-360},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2017-03-03 18:50:05.000000000","tz":-360},"subject":"Install python2-pyOpenSSL package on CentOS","message":"Install python2-pyOpenSSL package on CentOS\n\nPackage pyOpenSSL is obsoleted by python2-pyOpenSSL.\n\nChange-Id: I7953a8eafb2d9595f7c3758cddeb3f513246bfc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/93160e01ddbe370f64b2cceb25b888ac369d736e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/93160e01ddbe370f64b2cceb25b888ac369d736e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f02d890466859e2e14e6e030eee1b7620194a9c7","is_merged_in_target_branch":true,"change_id":"Ifaf2bf7f38853e57d9a1bddc20adf979f22d514c","change_number":440434,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
