)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~497246","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~master~I0bc0a10d1a9e19b424f0a330a5bee1b93cb4f261","project":"openstack/openstack-ansible-rabbitmq_server","branch":"master","topic":"optimize-gate","hashtags":[],"change_id":"I0bc0a10d1a9e19b424f0a330a5bee1b93cb4f261","subject":"[WIP] Avoid running pip_install twice","status":"ABANDONED","created":"2017-08-24 13:53:09.000000000","updated":"2017-08-24 20:44:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"57a8f80813074521fe979fc0d8a7b7c5e8214d70","_number":497246,"virtual_id_number":497246,"owner":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2017-08-24 16:55:13.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-24 16:55:13.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"fe2044a387375558f4ee0fdcbe34d94e34c77252","author":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"date":"2017-08-24 13:53:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5445835362acdd5d10ce7283e80790a67acc8699","author":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"date":"2017-08-24 13:53:33.000000000","message":"Topic set to optimize-gate","accounts_in_message":[],"_revision_number":1},{"id":"9399dfc76efaa3a3705b3d33952c72f024a61cac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-24 14:11:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial http://docs-draft.openstack.org/46/497246/1/check/gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial/3a72af6//doc/build/html/ : SUCCESS in 1m 18s\n- gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial http://logs.openstack.org/46/497246/1/check/gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial/f8a4eaa/ : SUCCESS in 1m 23s\n- gate-openstack-ansible-rabbitmq_server-ansible-func-centos-7 http://logs.openstack.org/46/497246/1/check/gate-openstack-ansible-rabbitmq_server-ansible-func-centos-7/32c0f49/ : FAILURE in 14m 15s\n- gate-openstack-ansible-rabbitmq_server-ansible-func-ubuntu-xenial http://logs.openstack.org/46/497246/1/check/gate-openstack-ansible-rabbitmq_server-ansible-func-ubuntu-xenial/d7b99dd/ : FAILURE in 9m 45s\n- gate-openstack-ansible-rabbitmq_server-ansible-func-opensuse-423-nv http://logs.openstack.org/46/497246/1/check/gate-openstack-ansible-rabbitmq_server-ansible-func-opensuse-423-nv/e64c50d/ : FAILURE in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c23a16011d2ac71024f89eb46fe8115c0592d824","author":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"date":"2017-08-24 14:22:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b927abd5052d92b31f30f1c3de37a89096bf8c6e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-24 16:55:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial http://docs-draft.openstack.org/46/497246/2/check/gate-openstack-ansible-rabbitmq_server-docs-ubuntu-xenial/5a202d4//doc/build/html/ : SUCCESS in 2m 07s\n- gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial http://logs.openstack.org/46/497246/2/check/gate-openstack-ansible-rabbitmq_server-linters-ubuntu-xenial/d449331/ : SUCCESS in 3m 00s\n- gate-openstack-ansible-rabbitmq_server-ansible-func-centos-7 http://logs.openstack.org/46/497246/2/check/gate-openstack-ansible-rabbitmq_server-ansible-func-centos-7/dd22b62/ : SUCCESS in 37m 16s\n- gate-openstack-ansible-rabbitmq_server-ansible-func-ubuntu-xenial http://logs.openstack.org/46/497246/2/check/gate-openstack-ansible-rabbitmq_server-ansible-func-ubuntu-xenial/41877ea/ : FAILURE in 7m 12s\n- gate-openstack-ansible-rabbitmq_server-ansible-func-opensuse-423-nv http://logs.openstack.org/46/497246/2/check/gate-openstack-ansible-rabbitmq_server-ansible-func-opensuse-423-nv/f7bfc6b/ : FAILURE in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57a8f80813074521fe979fc0d8a7b7c5e8214d70","author":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"date":"2017-08-24 20:44:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0005497588a285d4ba24e1af7be36a05197602c9","revisions":{"1acad962a922af8c9acfd135fef898ead590b60a":{"kind":"REWORK","_number":1,"created":"2017-08-24 13:53:09.000000000","uploader":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"ref":"refs/changes/46/497246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/46/497246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/497246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/497246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/497246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/497246/1"}}},"commit":{"parents":[{"commit":"7bb9dc14d06497a59cb930fa49b5e510041c064b","subject":"Optimize CentOS Erlang installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/7bb9dc14d06497a59cb930fa49b5e510041c064b"}]}],"author":{"name":"Major Hayden","email":"major@mhtx.net","date":"2017-08-24 13:52:15.000000000","tz":-300},"committer":{"name":"Major Hayden","email":"major@mhtx.net","date":"2017-08-24 13:52:18.000000000","tz":-300},"subject":"[WIP] Avoid running pip_install twice","message":"[WIP] Avoid running pip_install twice\n\nThis patch is a test to see if we can avoid running the pip_install\nrole twice. It is applied as a role dependency, which causes it to\nrun during the initial RabbitMQ installation and during the upgrade\ntest. That causes delays in the gate.\n\nChange-Id: I0bc0a10d1a9e19b424f0a330a5bee1b93cb4f261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/1acad962a922af8c9acfd135fef898ead590b60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/1acad962a922af8c9acfd135fef898ead590b60a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7bb9dc14d06497a59cb930fa49b5e510041c064b","is_merged_in_target_branch":false,"change_id":"I1bfcc96f353bd27b0004d93e250bb041eee48bdb","change_number":496376,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0005497588a285d4ba24e1af7be36a05197602c9":{"kind":"REWORK","_number":2,"created":"2017-08-24 14:22:25.000000000","uploader":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"ref":"refs/changes/46/497246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/46/497246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/497246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/497246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/497246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/497246/2"}}},"commit":{"parents":[{"commit":"7bb9dc14d06497a59cb930fa49b5e510041c064b","subject":"Optimize CentOS Erlang installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/7bb9dc14d06497a59cb930fa49b5e510041c064b"}]}],"author":{"name":"Major Hayden","email":"major@mhtx.net","date":"2017-08-24 14:22:18.000000000","tz":-300},"committer":{"name":"Major Hayden","email":"major@mhtx.net","date":"2017-08-24 14:22:18.000000000","tz":-300},"subject":"[WIP] Avoid running pip_install twice","message":"[WIP] Avoid running pip_install twice\n\nThis patch is a test to see if we can avoid running the pip_install\nrole twice. It is applied as a role dependency, which causes it to\nrun during the initial RabbitMQ installation and during the upgrade\ntest. That causes delays in the gate.\n\nChange-Id: I0bc0a10d1a9e19b424f0a330a5bee1b93cb4f261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/0005497588a285d4ba24e1af7be36a05197602c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/0005497588a285d4ba24e1af7be36a05197602c9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7bb9dc14d06497a59cb930fa49b5e510041c064b","is_merged_in_target_branch":false,"change_id":"I1bfcc96f353bd27b0004d93e250bb041eee48bdb","change_number":496376,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
