)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~567046","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~master~I02560d0fceea5a08be24f5c4b19a65d57c0f962a","project":"openstack/openstack-ansible-rabbitmq_server","branch":"master","topic":"add_bionic_support","hashtags":[],"change_id":"I02560d0fceea5a08be24f5c4b19a65d57c0f962a","subject":"Add Bionic testing","status":"ABANDONED","created":"2018-05-08 20:43:53.000000000","updated":"2018-08-06 13:29:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8660d91db334d6088a52866074d1450596d7672","_number":567046,"virtual_id_number":567046,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2018-05-10 01:50:37.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":-1,"date":"2018-05-10 02:20:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-10 01:50:37.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2018-05-10 02:20:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a44a2c64251ab258e230aa07fdc12bfaacbc1c93","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-05-08 20:43:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b0168b11830b48be99086871fbd702bfb054b4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 21:17:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/46/567046/1/check/build-openstack-sphinx-docs/3d93b4a/html/ : SUCCESS in 2m 56s\n- openstack-ansible-linters http://logs.openstack.org/46/567046/1/check/openstack-ansible-linters/f54cc9f/ : SUCCESS in 6m 06s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/46/567046/1/check/openstack-ansible-functional-centos-7/1fc8dcb/ : SUCCESS in 31m 21s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/46/567046/1/check/openstack-ansible-functional-opensuse-423/7405c1e/ : SUCCESS in 18m 05s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/46/567046/1/check/openstack-ansible-functional-ubuntu-xenial/950028e/ : SUCCESS in 24m 50s\n- openstack-ansible-functional-ubuntu-bionic-nv http://logs.openstack.org/46/567046/1/check/openstack-ansible-functional-ubuntu-bionic-nv/f8f2139/ : FAILURE in 5m 25s (non-voting)\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/46/567046/1/check/openstack-ansible-upgrade-ubuntu-xenial/eaa759a/ : SUCCESS in 24m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3df6a8c1de27310a00d9cf3367cec208c6659348","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2018-05-10 01:50:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5718a91e5504378b799b184a3e7e97f9e1a749d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-10 02:20:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/46/567046/1/check/build-openstack-sphinx-docs/673defe/html/ : SUCCESS in 3m 43s\n- openstack-ansible-linters http://logs.openstack.org/46/567046/1/check/openstack-ansible-linters/fd95c98/ : SUCCESS in 5m 34s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/46/567046/1/check/openstack-ansible-functional-centos-7/b4ffc39/ : FAILURE in 4m 02s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/46/567046/1/check/openstack-ansible-functional-opensuse-423/c31ff16/ : SUCCESS in 20m 31s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/46/567046/1/check/openstack-ansible-functional-ubuntu-xenial/5d3356b/ : SUCCESS in 23m 43s\n- openstack-ansible-functional-ubuntu-bionic-nv http://logs.openstack.org/46/567046/1/check/openstack-ansible-functional-ubuntu-bionic-nv/4887897/ : FAILURE in 5m 03s (non-voting)\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/46/567046/1/check/openstack-ansible-upgrade-ubuntu-xenial/73a3c89/ : SUCCESS in 24m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3eb2eb94f65d23a9f6bf284f42f03402184c3e38","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-08-06 13:29:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72e8db29d7f23556d5a34836aaf2210bd76b3c9b","revisions":{"72e8db29d7f23556d5a34836aaf2210bd76b3c9b":{"kind":"REWORK","_number":1,"created":"2018-05-08 20:43:53.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/46/567046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/46/567046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/567046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/567046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/567046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/46/567046/1"}}},"commit":{"parents":[{"commit":"c78e0b1d026ff34a53143fb897df0fdc8c5f8342","subject":"Do not restart rabbitmq when no version is changed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/c78e0b1d026ff34a53143fb897df0fdc8c5f8342"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2018-05-08 20:43:31.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2018-05-08 20:43:50.000000000","tz":120},"subject":"Add Bionic testing","message":"Add Bionic testing\n\nNow that bionic testing is added into the tests repos, we can\nstart testing it in the repo.\n\nDepends-On: https://review.openstack.org/#/c/566959/\nChange-Id: I02560d0fceea5a08be24f5c4b19a65d57c0f962a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/72e8db29d7f23556d5a34836aaf2210bd76b3c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/72e8db29d7f23556d5a34836aaf2210bd76b3c9b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c78e0b1d026ff34a53143fb897df0fdc8c5f8342","is_merged_in_target_branch":true,"change_id":"I523be647b5e82e6f088428bf2db24dc4cd2cfb53","change_number":566017,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
