)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~644705","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~master~I44ae90687f95de946eab8a72406bd2d0c553d7a3","project":"openstack/openstack-ansible-rabbitmq_server","branch":"master","topic":"osa-add-scenarios","hashtags":[],"change_id":"I44ae90687f95de946eab8a72406bd2d0c553d7a3","subject":"zuul: start running integrated infra-jobs","status":"ABANDONED","created":"2019-03-20 00:17:59.000000000","updated":"2019-03-26 18:03:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"34a3788063f7f72f1c715782f38c947a240d923f","_number":644705,"virtual_id_number":644705,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":-1,"date":"2019-03-20 00:54:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":-1,"date":"2019-03-20 06:42:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-20 00:54:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-20 06:42:26.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"}],"messages":[{"id":"239911480fcb9f12243701845763baa81f69ed8b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-03-20 00:17:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20818bd3bc4e38963585c207acd24b5ddf2a1413","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-20 00:54:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-infra-centos-7 http://logs.openstack.org/05/644705/1/check/openstack-ansible-deploy-infra-centos-7/3f487e4/ : FAILURE in 13m 51s\n- openstack-ansible-deploy-infra-ubuntu-bionic http://logs.openstack.org/05/644705/1/check/openstack-ansible-deploy-infra-ubuntu-bionic/0962dad/ : FAILURE in 18m 12s\n- openstack-ansible-linters http://logs.openstack.org/05/644705/1/check/openstack-ansible-linters/e29273f/ : SUCCESS in 10m 01s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/05/644705/1/check/openstack-ansible-functional-centos-7/3735532/ : SUCCESS in 33m 00s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/05/644705/1/check/openstack-ansible-functional-ubuntu-bionic/01f6ee6/ : SUCCESS in 27m 09s\n- openstack-tox-docs http://logs.openstack.org/05/644705/1/check/openstack-tox-docs/e8d083a/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"14cb038f84cd66a55705e62cf1c275cbf0d8c72d","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-03-20 06:42:26.000000000","message":"Patch Set 1: Code-Review-1\n\nSee comments in https://review.openstack.org/#/c/644708/1/zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"de7b4514d6e4080513f388b94f16e66621521116","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-03-26 18:03:50.000000000","message":"Abandoned\n\nabandoning for now.. I think Jesse will work a bit on this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43b3f05effd94c1ebc048297e2719af65bd245e4","revisions":{"43b3f05effd94c1ebc048297e2719af65bd245e4":{"kind":"REWORK","_number":1,"created":"2019-03-20 00:17:59.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/05/644705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/05/644705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/05/644705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/05/644705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/05/644705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/05/644705/1"}}},"commit":{"parents":[{"commit":"e907f84e1d66a6c96d4dce4642256a50d3eb0b72","subject":"Merge \"Do not configure openstack policies by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/e907f84e1d66a6c96d4dce4642256a50d3eb0b72"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-03-20 00:17:49.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-03-20 00:17:58.000000000","tz":-240},"subject":"zuul: start running integrated infra-jobs","message":"zuul: start running integrated infra-jobs\n\nThis patch adds support for us to start running the infrastructure\njobs which will exercise this role in the integrated repository.\n\nDepends-On: https://review.openstack.org/644703\nChange-Id: I44ae90687f95de946eab8a72406bd2d0c553d7a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/43b3f05effd94c1ebc048297e2719af65bd245e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/43b3f05effd94c1ebc048297e2719af65bd245e4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e907f84e1d66a6c96d4dce4642256a50d3eb0b72","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
