)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~714750","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~stable%2Ftrain~I0d985bd35fcbee85b7346c7b1f4f26306bfe97d1","project":"openstack/openstack-ansible-rabbitmq_server","branch":"stable/train","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I0d985bd35fcbee85b7346c7b1f4f26306bfe97d1","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-03-24 18:37:12.000000000","updated":"2020-04-06 23:07:41.000000000","submitted":"2020-04-06 23:05:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714750-1586214359026-c40b57dc","meta_rev_id":"f9d061a29c1694a143c0b5c5479aa676f0c2124b","_number":714750,"virtual_id_number":714750,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"date":"2020-04-05 17:55:20.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 23:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-02 20:28:35.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-03-24 19:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-03-24 19:43:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 19:43:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-02 20:28:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-05 17:55:20.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-04-06 23:05:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0a1fdf2918e63a0491d0bcc425ba9970b21665f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-24 18:37:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bb26accf46c8b15d359f98e600a5b4305df1539","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-24 19:43:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a99f9def8ee9695e239273101598534b7165ded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 21:34:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5e7c5ef3d063442f88ba078c623412df : SUCCESS in 8m 17s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/c0a70bef2aaa4754a3f8c2858d9913a6 : SUCCESS in 21m 13s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/3926710d1f084fe2b64a67d164437208 : SUCCESS in 21m 25s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a042ba55008c4375a00f3929cd141232 : SUCCESS in 20m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2c0336b0e2e5476596c8d5b841cafdcc : RETRY_LIMIT in 10m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e9d2dbaae05e416abd95d53994b0505e : FAILURE in 33m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44cf6e99508c4d85bbca69712b5be236 : SUCCESS in 55m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6c1c04f28fc04187bb53caf4da157012 : SUCCESS in 1h 14m 59s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8844fbd845d04352a6d3b695fdfdea51 : SUCCESS in 1h 29m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/711148b40080464ea9add4eec92e4386 : RETRY_LIMIT in 11m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d9f6b75dbd254e33873776fd51a88bae : SUCCESS in 1h 21m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/baacadd9428242e4a92bd2ed26c34d46 : SUCCESS in 1h 14m 15s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/112d2670445f4d3b948f4c51ee6cdea8 : SUCCESS in 2h 21m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f767fc8e79b64b06adfb4bddb4192592 : SUCCESS in 1h 49m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda81d6c8a4c4a8a98072f0542644e90 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccc714e7d6c14b68993847236d931227 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0f4b7f0f6e7aa38a0892cda1874416a7824079b7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-02 20:28:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"652bfdc938c31470d5f71ddc7c52abc59bd1f77f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:49:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/47278ad2c73f4b9e8c6d18c0c9205bae : SUCCESS in 6m 43s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/14fb4e010e284a388573f43f3c4ecedd : FAILURE in 13m 54s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2e40d2bdb93f44b48ad6101073f6ca2d : SUCCESS in 22m 48s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d1893d8c59448ffb0e1e1906b531941 : SUCCESS in 20m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b5b4aef72e0f4b80a4237e713e4d28a7 : SUCCESS in 1h 07m 45s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/66c7378e988c4f6d88977cc2389ac330 : FAILURE in 32m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f97221bfce9d43c4b2efce13a83edac5 : SUCCESS in 56m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/80ae0ad081614de9a33edc264c46974d : SUCCESS in 1h 07m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c1c639839a104aa5bb3401a30dba70ea : SUCCESS in 1h 25m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6c9b5b9654b34c4e8916019a6d473647 : SUCCESS in 1h 34m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/50611c4a599a4c66a93b48ac2c9683b3 : SUCCESS in 1h 33m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/416e8f86f40b4d2bbcd4606ea74bfd48 : SUCCESS in 1h 09m 55s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/51d09559fdac4b028e12a76bbee1478f : FAILURE in 13m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6d9f3c7dd28430983597296e2b49075 : FAILURE in 14m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20aedf71998b4adf8ad3438d4b67a0fc : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d279c44aee4b45b8be12a6854c307532 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"00e14b2cbba4b4372df6f03c936eecaee6c49409","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-05 17:55:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"54fbd605c0deaf9ece0d18683d690e31fddf1884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 20:12:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ae293780135b48b28c2a5554e2c5c037 : SUCCESS in 6m 48s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/c79390e7ca7f4f1587a2d89d26bb3725 : SUCCESS in 20m 45s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d8357cededeb417386df64ef0ebc84c6 : SUCCESS in 19m 34s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a34daa757694c9a81ee44b986fa86d3 : SUCCESS in 18m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e8827729cc7f43c880f042675340e63a : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/11f80c77b1f54e51a9317ad77c435013 : FAILURE in 30m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/764184d9525a4ab88f4b49eeb763b9cb : SUCCESS in 53m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/14c4c519c0ac4e338cba2756a39d9f06 : SUCCESS in 1h 05m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/46ace5e861ac4e51bf184bf3cdbf345a : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ed14612496be48028c54826d58ef4744 : SUCCESS in 1h 28m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7c6343825729413ab54eaa0138da8e97 : SUCCESS in 1h 29m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f8cfc6226a3419a866b9a12d4eb6c39 : SUCCESS in 1h 10m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d3faa462b871411bac12fc45e52b6387 : SUCCESS in 2h 15m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f29683a2ea0f4cbcb3164a3afeef16d3 : SUCCESS in 1h 52m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03451bfb4cd4defa6ee92b97826ef56 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9810a3e592364dd68f6b538048352957 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b176f464459713e2436ae8a742ee9a083e63a39a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 20:12:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c812338033ed00595eb504c1314fdb5120a40fbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 22:03:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7201490a024e4ac7b6603b12cb093ca5 : SUCCESS in 5m 16s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/ad5ad1c5ff75487180d650a171633d75 : SUCCESS in 18m 28s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/39b76762489e4f16ad3745d425f53637 : SUCCESS in 21m 35s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0a79ce02b7b4c0aa8df00a0365645f1 : SUCCESS in 22m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/618441f9e8b143daa5579005e5c96d37 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/508baf146a894cfd879f35a12eb1d958 : SUCCESS in 1h 10m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/99e3cb424ce3427891b6e2f29fceda91 : SUCCESS in 59m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/38fcaea02c6d444fa36b8c70e353bd0d : SUCCESS in 1h 17m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/75ffc188599f4b07b5047de7d96032bf : SUCCESS in 1h 35m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2c0423d6e099454e95785f45ab174b2a : FAILURE in 1h 17m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4448595cc7f45379761e932027d0efd : SUCCESS in 1h 21m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ffd2db91dd06401cbb6c505f5f93b8d4 : SUCCESS in 1h 48m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11d2b7442b12475aa8cfa8e32d433126 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0651b021db244b19be481bd5b1e208d8 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"578efc4e39de6eefe35d8ca09b009f69b8b37a5e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-06 11:52:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4eb152ced9f91de79e6032bedc55e508106eb87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 14:25:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/636dea7d2c724a58ba8aa65da9971644 : SUCCESS in 6m 28s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/1aa873f88f664643b5d92cf3933e63c8 : SUCCESS in 18m 17s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/701d86573f144cab9811f85584599644 : SUCCESS in 20m 36s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27362aa11f3840859ce204b9606cc8ad : SUCCESS in 16m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c9d782051ca64cff9c9501b7d7951179 : SUCCESS in 1h 07m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aff271bd69eb42d69c5c32916dfb2012 : FAILURE in 31m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/255c15b6dff04fd994589a9bd4364f33 : SUCCESS in 51m 26s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c225f2a967a744f5a24f5d8db550d5d0 : SUCCESS in 1h 09m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f044a5cc45d44f7cbae8b23d06cc988a : SUCCESS in 1h 10m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/34bd920208284ee3ba0856ff7574665f : SUCCESS in 1h 31m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e8e32392ca144f05b1f337028a4dd58a : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41d8634229ad41c8a936de435145aa60 : SUCCESS in 1h 07m 17s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e24b54323cf84f33b92b5e24acdd753b : SUCCESS in 2h 25m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddbf6c3bc5ea46d4abcc9d05f2f416bc : SUCCESS in 1h 52m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495da8225d0045a1bcdbf61c6bf3a0f5 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af6580024a1842f8985ee519f48aa91c : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"83de85b8f5d778a7d6295fda119d5b7428474d1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 14:25:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a23e234b415fc7ea021a4fd7cd95dc63b535b3ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 16:20:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3babbace4c5f412396a73fbdb7dfc744 : SUCCESS in 5m 38s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/ee90ef002e6e413eac64e38e8812da07 : SUCCESS in 19m 31s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/9419ae2d9c3d44cabd9b85a210d64049 : SUCCESS in 21m 30s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e82f00ecf2fa4a23964d510bb2d33340 : SUCCESS in 19m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6e70dd6e0fb04a0e8965dc0a39ca725c : SUCCESS in 1h 23m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1026e89dfcd945d9ad17c33313cb6461 : FAILURE in 20m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ff431803863b43599dfd7e8624b7ecd6 : SUCCESS in 1h 04m 18s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/207d3c27c1d84f4c9b49043352aeeab4 : SUCCESS in 1h 26m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/215a2ff134164463b34837a9ef98ca15 : SUCCESS in 1h 27m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/121db414db92459c8580d49c57172b5b : SUCCESS in 1h 29m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42d031c06089494ca85478f1b170ebc4 : SUCCESS in 1h 26m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3099f4c877f4a4097f0092f69beb64c : SUCCESS in 1h 47m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e70975139034cd088325110964a9167 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d48895819974706b2dce74eb2930e59 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"402e5a53e02735742a03844dda02f9b4fd1ef260","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-06 18:33:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4dd5261eb7e7b16709895452b90caa898c4b3c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 21:16:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/be53c55e566e4f5cb2938ee45c17a8fb : SUCCESS in 5m 21s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/d117b64b555543a3818f106e2360e358 : SUCCESS in 18m 59s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/90bd8bf70b9a463e98e46be1c978a234 : SUCCESS in 19m 27s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17d5da83eca74569a91e302668ff9744 : SUCCESS in 16m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/485c2ff53b1d4089af5c6d76406e35a9 : SUCCESS in 1h 09m 17s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f5d3143733e7484ab61054b90c1476bd : FAILURE in 28m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7779064fbb204306882c1606fbc02c45 : SUCCESS in 50m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/14c6a5981feb4d1781d4ab09d179b46f : SUCCESS in 1h 00m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6898222f326d423da3d2d5e492b4b042 : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/99159d6dcf4d4cf2a94a89ba858833eb : SUCCESS in 1h 29m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/285cfc55c3844d328c397a835ec22479 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4da1196c05f844258846f6607348b748 : SUCCESS in 1h 06m 32s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/909fe4fbfe2f408fac7e5b4ebf69be90 : SUCCESS in 2h 36m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e691e79f56da4a0294a6e79913a9353e : SUCCESS in 1h 44m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15c0a97365894b128843199df0912744 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/787f0ad000bb46889b7a41ab57dfd3dd : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"08a3ba9f6a3ecb28df6d0d9b034a034bcabc951b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 21:16:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eab4bd211190b6eb3dc88df8f5b79484834e379f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 23:05:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/85e0b552bc4b434ca1ea498f0b901221 : SUCCESS in 5m 55s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/f9db4bd735b94505ad450010628b9e13 : SUCCESS in 23m 09s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/bfe96a27d7fd4e6d9e010adde1f163b3 : SUCCESS in 21m 37s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b872fcb518c455dacc669f80844fdda : SUCCESS in 17m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2e137e1fb7a2414a8e56e5b9aad06e36 : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6750dce406d84939b739484c26552518 : SUCCESS in 1h 11m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/00117ded14454ae8bcc41abdbd687877 : SUCCESS in 1h 00m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9f8ef498eecb4bc59579300f3c09a613 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ac46df59f1d84eb698fbc6d3dfaf2c33 : SUCCESS in 1h 36m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/953d28f837f4415abf4a86ece392b0c5 : SUCCESS in 1h 17m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a38d8c028c14b11a308ee033093fa77 : SUCCESS in 1h 09m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c809f6a5420e42f68e9811e2b01e4180 : SUCCESS in 1h 47m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4532c07b3784736b722144b9278001e : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d183de048d34fa49c41c7dd922784ce : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ccacbe0043d682c644e8ecee71cbbd0afd4e7e89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 23:05:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e809d0012f769a504afe22337f33cb0e1ec2763","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 23:07:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ca5025b2bc4e54bb5831f6b4227604 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5970d7ea77e4b8c86db8d890199c960 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6921489c54274e045e049be02916f3e6b052157d","revisions":{"6921489c54274e045e049be02916f3e6b052157d":{"kind":"REWORK","_number":1,"created":"2020-03-24 18:37:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/714750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/50/714750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/50/714750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/50/714750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/50/714750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/50/714750/1"}}},"commit":{"parents":[{"commit":"165258301acf542c755e7802d9716b7ff882fffa","subject":"Switch to the amqp repository for erlang dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/165258301acf542c755e7802d9716b7ff882fffa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:37:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:37:10.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I0d985bd35fcbee85b7346c7b1f4f26306bfe97d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/6921489c54274e045e049be02916f3e6b052157d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/6921489c54274e045e049be02916f3e6b052157d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
