)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~734088","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~stable%2Fussuri~Ib7786bb6ce6ef5cfa5991ecdc370d7f8760a9550","project":"openstack/openstack-ansible-rabbitmq_server","branch":"stable/ussuri","topic":"osa-focal-stable/ussuri","hashtags":[],"change_id":"Ib7786bb6ce6ef5cfa5991ecdc370d7f8760a9550","subject":"Fix vars file for ubuntu focal","status":"MERGED","created":"2020-06-08 13:43:14.000000000","updated":"2020-06-17 15:52:50.000000000","submitted":"2020-06-17 15:48:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734088-1592408939962-c4077c53","meta_rev_id":"7ac38e913316c595b8eb336b926641b6d7cfc443","_number":734088,"virtual_id_number":734088,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 15:48:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-06-17 14:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-08 16:21:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-06-17 14:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 14:02:14.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-17 15:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5db17a45b3fd36e91009c2d3d2cdd6ca8010023f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 13:43:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a80c8c394c9a7a8251692a1efe12d7d0de19a17c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 16:12:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/283366951c8f43b48f3d46e30a81ed34 : SUCCESS in 6m 46s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/e5dcdfead5ca4347a9eceb28e18a5538 : SUCCESS in 17m 31s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/944c98b3e9e54ff5982075b74f93e705 : SUCCESS in 22m 01s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/7931a7ae08b94d4e97f828c8acda5825 : FAILURE in 11m 20s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81459b222c424f4682e2ff9013d9f8c7 : SUCCESS in 22m 46s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba8b94bf1a8f43fea1ab4eb8d755cb24 : FAILURE in 8m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b97fc12e0cf34952bb18464f96dc9db1 : SUCCESS in 1h 06m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c7336aed1854aeaaa7c034afcc661ec : SUCCESS in 55m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c3f68b7fc40340cdbe12b3b7c3e7127d : SUCCESS in 1h 06m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4a555b526b39479cbb7078c360d25ba0 : SUCCESS in 1h 22m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58472f71cbb84d8994e7ab944c10b0f9 : SUCCESS in 1h 07m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/14e26e61d0fe44b4b65fa778019ae5b4 : SUCCESS in 2h 23m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3f7ccab4b8444e997fca64e701384d9 : SUCCESS in 1h 49m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c412cfe74d4ec89b6980e3d168a604 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2f1108513f5b85be9b8dd988ccf477523defda00","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 16:21:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd773e5189bf492772f1ca024b0387a514240710","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-17 14:02:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a1ce08917018c44d1aa104a20ace31d721d876d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 14:02:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf37d49035c50ee764906a3dbd0f3c625f43a204","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 15:48:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"84953784029b56e71f41cc375605066b254063fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 15:48:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/15463aa3c38e4ad1a768df6159daf303 : SUCCESS in 11m 07s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7873c4b9deec41f8b26aa9f630108a95 : SUCCESS in 23m 27s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/eb8539d1be734cf8afb08525c21c2b27 : SUCCESS in 27m 07s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77d72bed9fa3434d8d8a15602d48965c : SUCCESS in 23m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7b15cd5f4a6d4a1a9c8f2652e1320015 : SUCCESS in 1h 21m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/670413f5da8347cbbd68d1fb67ec3615 : SUCCESS in 1h 05m 19s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/96ea8fbb132a42c8bf41d90a4762aa73 : SUCCESS in 1h 11m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/75b5c0266a3843ada040ff34e2bbab89 : SUCCESS in 1h 38m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7289ad118fc3465d97b22f317f061101 : SUCCESS in 1h 32m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7833b66fe41c44fb8521c09f54a6d050 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ff42b2ff2e4fddddd1e3ea1cbffc6e0e9a32d304","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 15:52:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/104040a2b141439faf450109e7066047 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1446ec998ee9e5691c92f41ad1955fb5574293d7","revisions":{"1446ec998ee9e5691c92f41ad1955fb5574293d7":{"kind":"REWORK","_number":1,"created":"2020-06-08 13:43:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/88/734088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/88/734088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/88/734088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/88/734088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/88/734088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/88/734088/1"}}},"commit":{"parents":[{"commit":"adde371e15b55d2a8d96c62b80240b6449daf037","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/adde371e15b55d2a8d96c62b80240b6449daf037"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-27 08:14:38.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-08 13:43:14.000000000","tz":0},"subject":"Fix vars file for ubuntu focal","message":"Fix vars file for ubuntu focal\n\nChange-Id: Ib7786bb6ce6ef5cfa5991ecdc370d7f8760a9550\n(cherry picked from commit 9a9946c12f51ae25b9f3b78e87fd9fb9ba5e0e24)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/1446ec998ee9e5691c92f41ad1955fb5574293d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/1446ec998ee9e5691c92f41ad1955fb5574293d7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
