)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~920999","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~stable%2F2023.1~I5211a30899f544a9f7e609e85551a92b245d25bf","project":"openstack/openstack-ansible-rabbitmq_server","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-06-05 15:28:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5211a30899f544a9f7e609e85551a92b245d25bf","subject":"Enable feature flags post-upgrade","status":"MERGED","created":"2024-06-03 09:09:47.000000000","updated":"2024-06-05 15:29:51.000000000","submitted":"2024-06-05 15:28:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920999","cherry_pick_of_change":919701,"cherry_pick_of_patch_set":6,"meta_rev_id":"0a0c38ecef9cd326e08e580a8e479caf13ab443e","_number":920999,"virtual_id_number":920999,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-05 15:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2024-06-05 13:08:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2024-06-04 18:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2024-06-05 13:08:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-03 09:09:47.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2024-06-03 11:41:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-05 13:08:07.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"2ca7e7a4b66d29c277c9fd1a4f2b722e70921f17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-03 09:09:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ccde193329fc2cf4c3df52ee217163b8cf62b116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 11:41:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b2b7395cc364ea881955a68fa1eaf65\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b99ad6f6e9a4386b5b285335f903980 : SUCCESS in 11m 03s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b43ed964f76f44959b63f0d4e37dd36f : SUCCESS in 1h 18m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ca6aa55d530b4317bddea61616bfc6ab : SUCCESS in 1h 29m 34s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9da6897d19ef4cf2a2ffbb09c17e0d1c : SUCCESS in 1h 38m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4673b4f1c3cd498da644539213f05d63 : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6179a6c4b70a4d989a9c6c90a32dbe7d : SUCCESS in 1h 09m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be8a3c4ff4bb43a098d73d47a0271826 : SUCCESS in 1h 16m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5b0bcbcf5e041bcb0a5f8984c293de8 : SUCCESS in 1h 13m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb72e3c6016e496999749a2e55676269 : SUCCESS in 1h 07m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3c6308bed9fe47b6b39e4f43bdc9aa35 : SUCCESS in 1h 23m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/74f4995ef7e54a44809affb12b8bdd35 : SUCCESS in 1h 41m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/291755119aa74173a310e0b1c5c40232 : SUCCESS in 1h 27m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae5b01b9caeb4b518520d65c518bbc36 : SUCCESS in 1h 35m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4754b27d5838459788984cec0de86e96 : POST_FAILURE in 1h 23m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/27afaaa01ef44447833ffd20161c0ff6 : SUCCESS in 1h 27m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/307b97a557ba4a629b49a15a047ac4d1 : SUCCESS in 1h 28m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/884f19c81f6a4433aa01a7fa65877253 : SUCCESS in 2h 18m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c722537605c4a81b6e2e6b6ad828f0a : SUCCESS in 2h 02m 26s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a6f2ae0f04ef4a8196fb2dc81fa56c29 : SUCCESS in 1h 39m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f89f7ce9841d427d9b840e1f0493ea93 : SUCCESS in 1h 23m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e804a4f5c1479385241af80dfe27a0 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b931be492fe4b6ba6307f10d45edf1b : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"90b2ad158e1b09710eb47fe46942bc2cff556b69","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-04 18:19:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ce9519f6f2b7c52bbf6c92835251ee8c3169f75","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-04 18:20:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"693f1ec8b9a9534df837d37844338f6166870253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 20:28:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b249dfd6444a4d0da45ef0aee785aed1\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f04eaeb6c2ff4ff9bdcca1e0dc58562d : SUCCESS in 10m 00s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e6f349abf864885a7a455b4b44fe61e : SUCCESS in 1h 24m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b1da568639974c669cbc012600488b18 : SUCCESS in 1h 23m 57s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4416bd4a769f4e5b9983e8ccc42d9064 : FAILURE in 1h 14m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a24d92edfe304d13be6f2dc3b8754b29 : SUCCESS in 1h 09m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c60acc3488be474ca90e45c7b69460a3 : SUCCESS in 1h 15m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e17a54ae8924e4b945ab7bd3e60198a : SUCCESS in 1h 09m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fb6fde1731f4728be674e8220cb0ada : SUCCESS in 1h 13m 44s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19ea36de82814308bba54cd77ca30849 : SUCCESS in 1h 24m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/38bcc9df43a44784bd4d744d660467b8 : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e48b5c2626a546a890ee95d3c6520908 : SUCCESS in 1h 28m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/14cc013901fa4a6e90b8bbfc83620dea : SUCCESS in 1h 37m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5bbe15adf4f400fbcd20365029593fe : SUCCESS in 1h 37m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d86b54dc5d7d4072975a36eae1ab8df0 : SUCCESS in 1h 43m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/599990c2fb8e483286e7f7c19f62d2e8 : SUCCESS in 1h 31m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54aab8ff0e284ffabaa47e8b946d06a8 : SUCCESS in 1h 39m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59d6db6f4e324f8db675b3b279a1985a : SUCCESS in 1h 57m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d4b1e63d5e0447b9427dfb23be32d1f : SUCCESS in 1h 57m 38s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e29e06bdcc714e5da8717de754c6ef47 : SUCCESS in 1h 52m 13s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0c598c0947c437f88d1a31e1bc0b83e : SUCCESS in 1h 19m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/492e8882b1804abe886fc9c26c7b8d1d : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7958393e3b6c480ab240a080327b1851 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5cb7d55e9b1cdffae9d63f6dce639eabf5d33522","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-05 07:24:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"822df34aea4419750a2c901fb2eebec0e4f3614e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 09:38:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60e0cad6199d4aaab2031187d95279de\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/887f59a4aa88431f82c6ad89b8cec7ef : SUCCESS in 12m 28s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecd0f764366347d997f9ca6d2b0ed0fb : SUCCESS in 1h 25m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2f12b0e0280449f9b52afa1c8b9f235 : SUCCESS in 1h 17m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7f65b3364f6343cab9317870eeaf1ca4 : SUCCESS in 1h 35m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9650409170634fd18df6f94ac52792c0 : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d51665c37f8040eebb8276b1c1ee5e37 : SUCCESS in 1h 07m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/617f749245a74393ba96d62db22f7c8f : SUCCESS in 1h 16m 09s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59c4a534ca0f4c8ca626fa90facfc722 : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6efa6b53d3b9463cb2cf057f0f9bd198 : SUCCESS in 1h 22m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2ab70e94134943049153a570d8880661 : SUCCESS in 1h 28m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/702e70d7e64245d4a56324d17941fe92 : SUCCESS in 1h 20m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/181a1ff8ba1940afb32791c420c31530 : SUCCESS in 1h 38m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a30e285aa7e468e8399dcf6fac46b6e : SUCCESS in 1h 32m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5810aa59328432492c353acb5fb7553 : SUCCESS in 1h 40m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d8c62b379c04bcaa609fb6cdb5979c2 : SUCCESS in 1h 39m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e34362f3e0da417bad040993ca20735c : SUCCESS in 1h 41m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b150b5a62b6846b0aa5655c3e8425123 : SUCCESS in 2h 05m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d35ddebbb38478dbdf5dac9c9bebbfb : SUCCESS in 2h 11m 02s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/46f8fb1bfb19435e9687fdcf5e8d0c12 : SUCCESS in 1h 53m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d77e564c83cc4932846366c228757a92 : SUCCESS in 1h 23m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95fc09f7a5234c34904de54dd9a0d5c1 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/527f3860ff6c4496a01153388f96f813 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9b4d9d3b95e1e5141f20253fc89be85426c8bb87","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2024-06-05 13:08:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"809e92cc9214759ce7f30fcbde16d1b26ba99dd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 13:10:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c1e26904e2e2a7139e7949d484a7afddcd8fee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 15:28:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8e094f39cae41bfa7490ead438cd86d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a9f67fc3af64d7db4b030962e65dea0 : SUCCESS in 11m 36s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67799fa8df8648618a21fc95b4acb1da : SUCCESS in 1h 28m 56s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9bf20c0103184fadbbc3a74cd5e4931b : SUCCESS in 1h 34m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0e85e023440c4d8293fc3a3c8672ba75 : SUCCESS in 1h 11m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60e79867a4ae4249994244f249926645 : SUCCESS in 1h 19m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe863d2b9f844d2cbe5d113e17533756 : SUCCESS in 1h 14m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdb39b75f5f844b3a2a3017ccd3d6a95 : SUCCESS in 1h 23m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b74a3a2b86a14bea8b56398871751a9a : SUCCESS in 1h 19m 09s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/97299e07e4c345cea89a75bb89d1c53d : SUCCESS in 1h 47m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4278e70b080b408eba1c24b8943eca09 : SUCCESS in 1h 28m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2fd2db5ac764474d9bf70ba2e10888e9 : SUCCESS in 1h 51m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10598a8797cc4045861b63b8bd6c5a7b : SUCCESS in 2h 12m 20s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a7f2c2c782a64d51b1e819b053368f33 : SUCCESS in 2h 01m 21s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eef31f7f55314fe88b172ab3af824356 : SUCCESS in 1h 28m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a187462a907a43a0ba5b1387ffc732ab : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/507a8c56c33c4c7683f4a47d6aa8a15c : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"76f77f7871c7e1371433785469dcfadd78c16dc5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 15:28:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0a0c38ecef9cd326e08e580a8e479caf13ab443e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 15:29:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdceced9d7ea474d81a502cf7eafd641\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c9e8ca236b4b77835e5f3d7d5b6e32 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a61e85039b5edd8b39f3585dfa11eabc47046812","revisions":{"a61e85039b5edd8b39f3585dfa11eabc47046812":{"kind":"REWORK","_number":1,"created":"2024-06-03 09:09:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/99/920999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/99/920999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/99/920999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/99/920999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/99/920999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/99/920999/1"}}},"commit":{"parents":[{"commit":"2e849ca843d9f23f5b7feabb73a1c4788c9140e0","subject":"Merge \"Remove RabbitMQ restart when changing policy\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/2e849ca843d9f23f5b7feabb73a1c4788c9140e0"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 09:06:09.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-06-03 09:09:47.000000000","tz":0},"subject":"Enable feature flags post-upgrade","message":"Enable feature flags post-upgrade\n\nStable feature flags must be enabled before upgrades to ensure\nthe service can still start. These additional tasks check for\nany disabled feature flags and enable them. This is done\nimmediately after upgrades so that systems can run with them\nenabled striaght away.\n\nCloses-Bug: #2038818\nChange-Id: I5211a30899f544a9f7e609e85551a92b245d25bf\n(cherry picked from commit 5abd7b71ba87c3fdc5d1e1c790347e8da0a229cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/a61e85039b5edd8b39f3585dfa11eabc47046812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/a61e85039b5edd8b39f3585dfa11eabc47046812"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
