)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~931800","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~master~Ie453f327850f9c9ae54823c17b0292517fa56a82","project":"openstack/openstack-ansible-rabbitmq_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-10-18 09:57:59.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ie453f327850f9c9ae54823c17b0292517fa56a82","subject":"Proceed with installation/upgrade even if cluster not healthy","status":"MERGED","created":"2024-10-08 14:55:13.000000000","updated":"2024-10-18 14:50:43.000000000","submitted":"2024-10-18 14:49:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931800","meta_rev_id":"a600f53a085c8d89df9c7196d602185c12c24028","_number":931800,"virtual_id_number":931800,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-18 14:49:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-10-14 07:58:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-17 07:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-17 07:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2024-10-08 14:55:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 13:43:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-14 07:58:37.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-10-17 07:29:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"6124125a011a59c8d014e3fdddbbfbc5505399c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-08 14:55:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdb7d8fbaaa2d58de3ece8d6ad313daf4c0f79f9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-08 14:55:48.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13ba5aec3eeddb4c3fab3e07eea8169ca670d52a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-08 15:02:02.000000000","message":"Abandoned\n\nshould be only on stable branches","accounts_in_message":[],"_revision_number":1},{"id":"2bbe1cb9269532e9122eb7a5e4055a18876df6de","tag":"autogenerated:gerrit:restore","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-09 11:14:09.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"4ca6ddbd673cee9c56a3a7e3a288f4d1990de423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 13:43:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6be1bf59514479dbb01c79c55e57023\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b702f7a3bcc408cbde7f1a4ee7195b5 : SUCCESS in 6m 48s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/73c5d3e6008044cdb2b9ff70e72c5216 : SUCCESS in 1h 12m 55s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51ed166e975a4dc6af8ed524654b0b6f : SUCCESS in 1h 17m 14s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2873c3ca79b44bb4a77f8b19170679ce : SUCCESS in 1h 32m 02s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b95fe6d8d3ec423d9f9e2aeeb9663ff9 : SUCCESS in 1h 06m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15bf9372ed604f6cab07e0cf8d463f7f : SUCCESS in 1h 02m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db9df6c96cae47b2ada4378395e70aed : SUCCESS in 1h 08m 22s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20da81ee26b942b191e02073c1bd7b19 : SUCCESS in 1h 18m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b3728a8b79d492a8e01d43e0226cd35 : SUCCESS in 1h 08m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/917e335939da43e68755caee180ac29f : SUCCESS in 1h 33m 47s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec744e3d0dea45c0b68e78e0d4526b90 : SUCCESS in 1h 21m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bbc7193cea0140d49bfd3b59759773a3 : SUCCESS in 1h 14m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/daa85002b70a410697d5baeaa097971e : SUCCESS in 1h 17m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67c8bfb51edb4935b0c7246a4984a0a5 : SUCCESS in 1h 21m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e70cc470d44b4b00b70ef62603d1cc79 : SUCCESS in 1h 11m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e782158d43e4f64af1483b671a3c6c5 : SUCCESS in 1h 17m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fff3c132eb094c3f9cfa2f9b0c417068 : SUCCESS in 1h 15m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5cd16fb9ddf24dbb90c55c9a2d0f2371 : SUCCESS in 1h 14m 30s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8e7c03ccfa8408bb89864ee3455a78c : SUCCESS in 49m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cccf370f28f24ff4b71e36039bf857f0 : SUCCESS in 52m 35s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdea3da0855b47eb919360d33176db58 : SUCCESS in 48m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c99ffa00dd4f458aa11fc1babda73fc4 : SUCCESS in 1h 52m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4259d5d880e44ed284df1037d70c1129 : SUCCESS in 1h 49m 40s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2dde4d02408c4c179eb2247d85837bab : SUCCESS in 2h 12m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2cb9df4a529478b913c14a06b4764ba : SUCCESS in 1h 54m 42s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d658592348e4495a9a18fcc2ac813168 : SUCCESS in 47m 34s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5223b1cba65b4fbdaf3471364a94c98c : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/820f12c9ca84471fa88336b59a411b14 : SUCCESS in 1h 01m 55s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/55f1aa2770a94b7685a5a2cd9db60873 : SUCCESS in 1h 51m 43s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e15ab6214852401092f00c107068f1a2 : SUCCESS in 1h 39m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4157cd046194492e974a57f022939dca : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"087717960aa5939386752a30ce7c714624fdce09","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-10-14 07:58:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a8e80dede8751cb69e15c4bf0872123e0ba1178a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-10-17 07:29:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7eaeba4152c27667a84736719e4ac2f6cc9ba39d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 07:31:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"930c4f605b9c3e778346bf8969ce8264cad7642e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 10:35:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffceb57e8bef4d05a3414824a0593311\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55143e6dae05461596ba968de12db109 : SUCCESS in 5m 45s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8e02f77825834e52a4039cad09150587 : SUCCESS in 1h 23m 41s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c42bfe99e7546c8a257ea1eb513d4ce : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e4ce8c9cd31e44469655711f57e76c32 : SUCCESS in 1h 13m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2ca80015c7a4a0dbc73cfc60acd8786 : SUCCESS in 1h 27m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ede6b514771408fbf0a6852e1c25240 : SUCCESS in 1h 40m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da14110118ff47cda6ce6a49cfb271a0 : SUCCESS in 1h 57m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/187f0a67d6084bbe96b46fd5fb3eb821 : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3578b6513414bbbae33fda2f6a5ef5e : SUCCESS in 2h 20m 42s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d1a057961b24c4699eac8574797337e : SUCCESS in 1h 32m 22s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/70e65e35090f40cdbcf49fd1116d365d : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dc8c2cce722b490b97f1834d4a87a96c : SUCCESS in 1h 12m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38308d5ad87646498aa438bda7e44233 : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12bf7e067d444754b4b3187a29ab5c8a : SUCCESS in 1h 11m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a628aa2d1ca4ba2ad3d6371153192e8 : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff70b1e8e2d842229074bdde16879f60 : SUCCESS in 1h 16m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3193091b47934db891fab017e8a1495d : SUCCESS in 1h 19m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7eec8e5f47bb41a28259824dcd7a33a4 : SUCCESS in 1h 02m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a2aa10d274c452f974e2ea67df5ea22 : SUCCESS in 1h 23m 23s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9bc16eb4eaf4ec09524e6260e81b912 : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5ab154c9d550449b9fdb1dcbde445c78 : SUCCESS in 57m 12s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f4139202d74436ba22f0ff758c8bfaa : FAILURE in 1h 24m 49s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/23ca96e2abe74e5d9b919322f9070851 : SUCCESS in 1h 53m 53s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90e8d8d7224842bd9d8d799b8d338141 : SUCCESS in 1h 55m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8157e715147d4160a7aa50a6f891c374 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"cfc5a882f4e0d99d1b785433954dbb40c51cf810","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-17 11:22:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a46d3fa42c053119f23294ce4c5775b6f6bab31c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 13:26:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d87ffa2d07a1481393252533189b9073\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fa239169def46d7a928448ad3f0c708 : SUCCESS in 7m 59s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5c6d468a73d456fac9cc2dbc74af593 : SUCCESS in 51m 48s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/479b3e5e1d48444a84373a33efe6afe9 : SUCCESS in 56m 14s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c5cae306a8a24cddaf4702fae6cef32a : SUCCESS in 55m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/50558db4a5e44b479cd1c8a370921098 : SUCCESS in 56m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ddd02cc2b204955a77fffdbeb385e5d : SUCCESS in 48m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9187163491f249d499be6c884654f5cf : SUCCESS in 49m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd87b0a95a6b4d23a342db8c0f9a20d4 : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42da1c4990e84e3eaf0493939eb75058 : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4aad7b2b001a4782aaa70a28ec80cc4b : SUCCESS in 1h 12m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83a852bf96b641b5994255f917aaf4f3 : SUCCESS in 1h 12m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/612ee917c60f45eea0631a62cf8a11b4 : SUCCESS in 1h 06m 03s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c424437c3cf422498a4a21efcbfd6bd : SUCCESS in 1h 16m 24s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5e33a99718a54db9a7590ab7bf5498fa : SUCCESS in 1h 16m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee38b3187dc44724a02f9fb6679efe42 : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db357850f03844149a48c85f4a1c0c69 : SUCCESS in 1h 07m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a160882b6634e61b17a1a035c70c985 : SUCCESS in 1h 23m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/642885f868f84b2892f53bf70db0f6fb : SUCCESS in 1h 14m 49s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/023e59016f1345e5baabb0f0f135d17d : SUCCESS in 1h 15m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfdeebb28f7f4af286c9b18e3d644b1b : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56734cb9aac14e5d89d29071cbff11c1 : SUCCESS in 1h 03m 11s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/43834e83fed948389c809df7aaf54958 : SUCCESS in 1h 42m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76dfb36d6b01472aadedb399aba1e557 : SUCCESS in 1h 51m 12s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ef7111e8278443f933b79538bba40a4 : SUCCESS in 1h 49m 55s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/120db9ea0a2f4e5eb813d0682695d397 : SUCCESS in 1h 54m 22s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1f24cdc6e138423a9b6941ea0756e25a : SUCCESS in 1h 06m 30s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/793a2b57e5504e4d8f46bdc173338e99 : SUCCESS in 1h 07m 13s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7de8bfd10ab84ce084f62303762559e2 : SUCCESS in 43m 47s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f9f232f0dd2f43d4b46c78746b284916 : SUCCESS in 1h 12m 30s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82bf2a7ef6ba4a16b0ff05852f1b532f : SUCCESS in 1h 03m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ad7e0f7e634ede95907dce74c74c73 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0c35ee675df116dc59c7bb8db4fb05649378024d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 13:28:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37947fcdbe78aaa56b87b70f08cd61b8617c6b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 15:37:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a04bde77c954b31a3ba40ae0e09ad0c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1e4a18e3f9741d0a4be4ee88fe2ffe7 : SUCCESS in 6m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e483c4c9f6540439f1facc8df7591f0 : SUCCESS in 1h 46m 31s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2c0f174e036f434e8efcd7183dec3dd9 : SUCCESS in 1h 27m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/48e86257b45e42479bea1fb97d166b08 : SUCCESS in 1h 11m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a1f56b4b16b4a84be88e91637096749 : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95666c28f29b4fe3966b71071578ba35 : SUCCESS in 48m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ebe7bceed7c4ff7a7e936ea9b65fb13 : SUCCESS in 48m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b4e6c3c97294ff580213ced2200fbfb : SUCCESS in 50m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5fd1b6c2eac49608a01be834f579af4 : SUCCESS in 1h 08m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d7d77959e7a45eebbb16dfb89c75382 : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d46865f405cb470f9521e6f118f20931 : SUCCESS in 1h 17m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81d09a42eeaa4c02a5dd2b3d0d29bee1 : SUCCESS in 1h 12m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e517c3bd9716456696bde5f603c12e7d : SUCCESS in 1h 13m 03s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4151e64ad97465f8e114e61dd3f27ca : SUCCESS in 1h 14m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65aacb4395b34e39ac68404a31e273f6 : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc6563f74d6e4eb08ce5de6f35390df8 : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b165d2dcd064c4ebb8a28afdae0ff4e : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3500daa106d64585949c899dd22a96d3 : SUCCESS in 1h 10m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b181d2c931d749338090ef46fc0ef028 : FAILURE in 1h 44m 50s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ea1946ca6ead49258326f71bf96f81af : SUCCESS in 1h 07m 35s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/733bef4da19e49d3b2f4f55032463c17 : SUCCESS in 58m 14s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9abe0bd09b664738a02035ac5b5b3b63 : SUCCESS in 1h 07m 09s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f5ad170e7834dd9858aa9b31e00a729 : SUCCESS in 1h 59m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/709c3135211243efae72c65a44e81e2a : SUCCESS in 1h 35m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3efd0ec2b443ac8c16f1fa7f05106d : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"daa28d97a5ebab07f45380820ff4abc3ff53851b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-17 16:37:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5992caf378ba8090c60bd1887e43dd23cdf17911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 18:51:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7985de1042f34117afaab89ee4e3e335\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f84bac807cc4de8b87ffcf74a10ed47 : SUCCESS in 6m 09s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c84d92d83ac04be499f3d7de52e80dd0 : SUCCESS in 1h 09m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2dacec14754447ada6e9e67953b94432 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f3f4c38a8ed64ac6a111681d72337db4 : SUCCESS in 1h 18m 56s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/76b906a192de403480d301591ce57295 : SUCCESS in 56m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ff4ee05e0424e018337a5efe953e7bb : SUCCESS in 1h 07m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f1b7ff7817d4fc899f8cda65f11036c : FAILURE in 50m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d463a500b9724d5c98173e4eeea18995 : SUCCESS in 1h 18m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7afd25b2001141528533914e70e74796 : SUCCESS in 1h 06m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d69e454c74f44f82ba6057fdbbf7eff4 : SUCCESS in 1h 09m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46e94af9ffef421ea9e96940b0240803 : SUCCESS in 1h 20m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/667deac8e7d840ffa0645c49529f2dd7 : SUCCESS in 1h 21m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7610c8a841364a719361b0e1fca6a063 : SUCCESS in 1h 12m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0685f3a621ab4ababcffd9617c883a3f : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0afe1407b675401fbc6615fd139ac593 : SUCCESS in 1h 15m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b99d381234524a589e4461cb4ffeee5c : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/08f0535d682744c4b81e77459b3b42a8 : SUCCESS in 55m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c3c56118e114061a4ff1400c28d6426 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c64368ede8bc4a27b75230a7ed3af419 : SUCCESS in 1h 20m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4d86e02b4cc41c4a33c0f6911baf6ed : FAILURE in 43m 42s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44cb27642e8045e4b35710436aad0801 : SUCCESS in 1h 18m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0367d4b9eedf49e78c7817a00f0a20d6 : SUCCESS in 1h 16m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac6fd327a4554266a8c7175859887a3b : SUCCESS in 1h 50m 13s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf783697d983419a9c7da8e9bd4ff526 : SUCCESS in 2h 06m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bf3f875f01f4338a772b1ba5d9df160 : SUCCESS in 1h 34m 05s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2340641a4dcb4ed5b146cedd166029a1 : SUCCESS in 1h 05m 23s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ac262d1f80854f80ba2431f3705e551b : SUCCESS in 57m 10s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a3d86f9ff27431483be49797ed39d65 : SUCCESS in 1h 05m 07s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b01c763056a34158883b910021227377 : SUCCESS in 1h 40m 32s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a62a6b2dc00d4397aca950dfb583547f : SUCCESS in 1h 24m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c3e7ee24a1461192528d5a2e512dc8 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"dcfdf765db4ba21314f9736e87cf798f5cbdfd61","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-18 09:57:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54afe62925db17c15945b2105fb4d22471e746c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 12:20:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b36417f21ae45819d597979b1d27569\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c8b3ee3b65447d899f65beacbdd5397 : SUCCESS in 7m 06s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2dba483fb6e942bfa968192a716e07b3 : SUCCESS in 1h 16m 07s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5cb5aee2aab94f88bd3944246046ef76 : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e09928c2a8c49c1814e2622ece271bc : SUCCESS in 1h 25m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/df1c5f6b28034b8689bcd4745e6d9a13 : SUCCESS in 1h 13m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3728bb08331e46e38287fe5cc851566c : SUCCESS in 1h 10m 16s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3456b2050714644b04bfa170715f8fd : SUCCESS in 1h 09m 22s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f5651ed2be640169c7a5b6c27064066 : SUCCESS in 1h 20m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb60a4e93d474e158d8d7b2cb22eadef : SUCCESS in 1h 15m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c4f718270b4405d885e9a73e8eefe26 : SUCCESS in 54m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b709d88970ef47e183d9862af8f6b1d9 : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5dc2d2f323d84ee1976889fa0c37f644 : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d24a2ba40244e49b810d05f74d47f43 : SUCCESS in 1h 28m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/23db0d9bd52542d3b134ea62979bebfd : SUCCESS in 1h 24m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ed1b53733d94ca5885ebe6b419fe2ed : SUCCESS in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e20dc4be3ac74b51ad822b76061a7b2f : SUCCESS in 1h 20m 33s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26380881cf33436e9f05042f07cf2547 : SUCCESS in 1h 23m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c21f5f15b06a41b4bf2d435e1f599f31 : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84e39cfc3e154f3cac2e0f2076490c34 : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6701db119a654d789b51c15d0e60016b : SUCCESS in 1h 16m 58s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3a438da418b4476a21651f0e8296a80 : SUCCESS in 1h 19m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/646f6947a9714a29b98308b10dec425d : SUCCESS in 2h 13m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c102c3873a784dfd9190ecc448ffea61 : SUCCESS in 1h 48m 53s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f2a596691bc2473090dd05fcb8b14721 : SUCCESS in 1h 58m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa140baeb4bb41d0b3e4d0fd8d24d419 : SUCCESS in 1h 53m 57s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/356344b789904ec898b5c77159825c52 : SUCCESS in 56m 34s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9d7325120871447ba18849bf3356c05d : SUCCESS in 1h 09m 05s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77013949e2c74cc092a342eb14a9b664 : SUCCESS in 1h 05m 58s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b37ac957b9394401bffc5137dbfcbec5 : SUCCESS in 2h 03m 33s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de3e2893d70c4779bcb3de08a558073a : SUCCESS in 1h 34m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb86b8e41234479adfb04cba7b0952d : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b501718624051bdf742faf11287ee6bd5f1fa3fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 12:23:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5faedb5d0650e010cb8c241c530055282201f6be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 14:49:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e876461b5fbe4ae68bda227527cefe76\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa48c5ad1bbe4d3a81a08af9aa34d826 : SUCCESS in 7m 42s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/767cec5e97164fe2a884183c069ebb06 : SUCCESS in 1h 23m 41s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/931ee6450d044655bb1a70d3cf70830d : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a4134ba5938544e79649f843df89cd65 : SUCCESS in 1h 10m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab2c076e35f14ffea07f1adf9bcb5fd9 : SUCCESS in 1h 09m 55s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1026ee379a1341f8b340a59944f41aec : SUCCESS in 52m 13s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/300a14ca5ef748e9a7d06758cffbe0d3 : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67515b27239c48c5abe49acb51817fd4 : SUCCESS in 1h 13m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f19176b7e74b46039284f9355ba6dc4a : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e73621b4a5a4f00a86cceedeff640d8 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/66c978f7feb2435188d017c6ccbc1f83 : SUCCESS in 51m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/02aa2df90ca94ecc956d6bf08feb119c : SUCCESS in 1h 10m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/518f926172f34ad3b113a15557eadfb3 : SUCCESS in 1h 06m 06s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66984d739df849b9afac0429f459084c : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdf51719b2c34432ad7b818d59f61a6c : SUCCESS in 1h 19m 27s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f13e504eb3644261aa4536c08ada658e : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/756720b8e6ee4d95a3adf36962796f23 : SUCCESS in 1h 20m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24e90c8794c54def88951b72f2248f35 : SUCCESS in 52m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/546e8a7eeac84a98bb3a7dc03179030e : SUCCESS in 1h 21m 52s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67a0de2ce8f94968ae1da656019d2853 : SUCCESS in 55m 57s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8bc74ea0f43943f8b1489246b0ec166c : SUCCESS in 1h 02m 46s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/174938c1288d4140bae55a1d23dfd8d6 : SUCCESS in 1h 10m 08s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/34de810c2f3e446d93fc76750a0bfda0 : SUCCESS in 1h 53m 34s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2153eedd50364cf095a3d83bfa5324bc : SUCCESS in 1h 37m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7edc59251df44406b8cd9a275fafd333 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3adfc1f4aec776cd8f1fb6beb601659f86afd870","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 14:49:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a600f53a085c8d89df9c7196d602185c12c24028","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 14:50:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e71608fcf8824fe194b5db087e24d613\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6637fb49630b41e5926c1318925ec91b : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b4a766db69a011edc9603e55f40565b08c82d3b","revisions":{"6b4a766db69a011edc9603e55f40565b08c82d3b":{"kind":"REWORK","_number":1,"created":"2024-10-08 14:55:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/931800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/00/931800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/00/931800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/00/931800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/00/931800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/00/931800/1"}}},"commit":{"parents":[{"commit":"de2c7d19d36006dc4d7456c787f37ca943510cfd","subject":"Revert \"Add retries for feature flags check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/de2c7d19d36006dc4d7456c787f37ca943510cfd"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-08 14:55:01.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-08 14:55:01.000000000","tz":120},"subject":"Proceed with installation/upgrade even if cluster not healthy","message":"Proceed with installation/upgrade even if cluster not healthy\n\nAt the moment rabbitmq_upgrade flag is being used on unhealthy\nclusters to repair them. However with change [1] we\u0027ve introduced\nrunning this task standalone very early during the playbook.\n\nIn case cluster is unhelathy - there is no good way to repair it anymore\nas\n`rabbitmqctl cluster_status` might fail right away with.\n\nSo let\u0027s ignore failures at this point and run feature_flags only if\nwe were successfull on fetching the status.\n\n[1] https://review.opendev.org/q/Idf33b6810e1afb3a89dad5e65a48e86ac0a58a67\n\nChange-Id: Ie453f327850f9c9ae54823c17b0292517fa56a82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/6b4a766db69a011edc9603e55f40565b08c82d3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/6b4a766db69a011edc9603e55f40565b08c82d3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
