)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~931974","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~master~I222b9de3ca531724b05c7fa2cf032a3271055d64","project":"openstack/openstack-ansible-rabbitmq_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-11-20 10:36:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I222b9de3ca531724b05c7fa2cf032a3271055d64","subject":"Move RabbitMQ restart to handlers","status":"MERGED","created":"2024-10-09 16:28:48.000000000","updated":"2024-11-20 10:37:43.000000000","submitted":"2024-11-20 10:36:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931974","meta_rev_id":"fad0cf49a89ede17d39b1e2d8fe77ca4e1527a23","_number":931974,"virtual_id_number":931974,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 10:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-11-20 08:33:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-18 10:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-11-20 08:33:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 17:55:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-18 10:23:30.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-11-20 08:33:51.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"fe1cf59eb1c1b4b706fa3986a8672cdd1230199d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-09 16:28:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0da73012a1d98eee6c88d8e6200287ecb9cf5d5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 17:55:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85b8c241461c4236a1afff933e441175\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2d1d261d7884b3991e8dbc647dc8b21 : SUCCESS in 5m 59s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b048bc2dce24436b9828e906fb47a46a : FAILURE in 44m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a94fab965214401b7a0218e52929ad9 : FAILURE in 45m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6ca82aabef164c0f8af72687fb46ca94 : FAILURE in 43m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ffe441fcfc114296a55a704088ce089e : FAILURE in 37m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29509e81eaf6468fa9383817beb8c8f3 : FAILURE in 38m 02s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19c1f6ac08844d17b1fd14061e4f65f2 : FAILURE in 37m 22s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b74873d0a5524369a5127a8f0e05a0ed : FAILURE in 38m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59d1ce87135a49719db8e45e582463f4 : FAILURE in 45m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9ac7340e48947ed9abc406bb859cfc7 : FAILURE in 38m 41s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8008e4c0b676469cacb686eaf3101bce : FAILURE in 46m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c52787620b9e44b6ac717709af7108f5 : FAILURE in 24m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4834688c10fe4ce08913357a062a5822 : POST_FAILURE in 23m 07s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/596b68fff88141748fc1204a5d8f59a1 : FAILURE in 24m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff23512e67c7497da9517767021d20b8 : FAILURE in 21m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9590016cc6846b686f6bc99a1ab347a : FAILURE in 24m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/020d1056337345b98b1bb8543ad4ef43 : FAILURE in 25m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/821a4542903d411e8df750888e92af21 : FAILURE in 21m 26s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fb49bcd3d2449c8b765a29feb251760 : FAILURE in 14m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dfcfc3d43964029a7d1a4f3b213dc2d : FAILURE in 22m 33s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca26bb085ffe49acb0232a21771258e5 : FAILURE in 20m 01s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f72bc81b29c543f18b038aaee9a54976 : FAILURE in 1h 13m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/330a5df596a444faaa8703cafa4882be : FAILURE in 1h 13m 10s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/18e507a607e947a3bae67db117ccb764 : FAILURE in 1h 22m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f4ca66b4a704bc68e6c63a14342839e : FAILURE in 1h 05m 21s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/414df5bfe68f4af0b93a05073b0da688 : FAILURE in 45m 16s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3f35c4eb31a34656bc23a7a79ed26af9 : FAILURE in 28m 25s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fe10c9850dc430e820e6f207e0b9af0 : FAILURE in 35m 41s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ffda68da70fd496fbd6d05097fcada7a : FAILURE in 56m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d37f56f016d1442b865765afd24ff681 : FAILURE in 1h 03m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31d5affdede48d88ea7acd134dcf860 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9cab80f08650589623dc9866fb863a6de354f075","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-10 07:04:11.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8d625a34fe15b9337263fb429ec0b11761774524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-10 07:04:49.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"9c99551e1a7f9ca50ac3ac347f81ab5833fa472d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-10 07:18:57.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7972b1faf6ea0a8ed7927485169008416f8dc2b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 09:54:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/befd88d4ce234526aa25db4cf73493a2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40c97009a3d147d791066a917897671e : SUCCESS in 9m 18s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b419771ecc6f477fbd1e26e2623866f9 : SUCCESS in 1h 20m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f77f95387bf94774a5c7916bb5919fbd : SUCCESS in 1h 22m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac9e8382bba2418fb45b8338e6cb4cea : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/68721cafb7944970a95180799c2a991f : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf85f0b565534c4599377ce08ca67459 : SUCCESS in 1h 13m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b30a4fcbfde34992bbefce5e8735c6ad : SUCCESS in 1h 16m 12s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e38b8d749834278aacf6cd982e49fc3 : SUCCESS in 1h 12m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85b1ab14b70e4e308a57652d3d8f271a : SUCCESS in 1h 08m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4feeefd0d434ae7a728387abef713a2 : SUCCESS in 1h 24m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/231ea386037049a3aa4399544ae39cb7 : SUCCESS in 1h 20m 49s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/212a72c3719047f29a71c913bbd83d4a : SUCCESS in 1h 18m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe44b3dcfac64be597207254fb092ca3 : SUCCESS in 1h 22m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6726e58cd83841e1a080ab9187c48ad1 : SUCCESS in 1h 26m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bb8d9eebd564cc69c6323be81d74ac7 : SUCCESS in 1h 09m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce2c9e7f83d64612bbffeba7de651fb4 : SUCCESS in 1h 12m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/660cadb286454b08b33b9b6179bd0bf1 : SUCCESS in 1h 24m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd1252ea179d4c8cac74600a93c61d45 : SUCCESS in 1h 05m 31s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fa088f2345c45b09439195c48b5edaf : SUCCESS in 1h 12m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f93eef43c9b4c6b9de20c4c74a61306 : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f42069951a14081b615685e790ba05e : SUCCESS in 1h 14m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d61180a74e924ceeaeecda86e77e4cf9 : FAILURE in 1h 17m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a7cb80512a64311b0de902040dad763 : RETRY_LIMIT in 2m 59s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aea5ab920dbe453da7e776e99a72a515 : SUCCESS in 2h 18m 48s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc4ab3bca46949b0b62bb50c588b48cf : SUCCESS in 1h 48m 11s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/47b898722b4b48279a7392b1c98dc2d8 : SUCCESS in 1h 12m 52s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c6f56946cb748ea940207e081f00d9c : SUCCESS in 58m 08s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32e000ec80824b139ab73a513dfa3d90 : SUCCESS in 1h 09m 24s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b2b4f3c4bab24d7ca8393ad2a6f8be4e : FAILURE in 6m 53s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fda0fe7430d47c5b358707434e57923 : FAILURE in 1h 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfe6e2873c2448f961edbdc9ac3da43 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":4},{"id":"82188ccea8990d7a6a9b203cf9f93ec2add75373","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-10 14:55:30.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f79d88e7c40144adcdea58b7d1e38e13de59711f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 17:02:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c06a3b1dec54635aafd0a932d23c1c4\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edaceb82cb4442159aa5dd09f28127dd : SUCCESS in 7m 42s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/16851978171d4df28a13829a27b8add6 : SUCCESS in 1h 20m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4682c141f0214945928c9c2778f9a3e9 : SUCCESS in 1h 20m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/998c25a4fe7c4618a15f9f62d70d5db8 : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/78d9112bed774f73b2fd278e9ef9bfdc : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d6ef7fac1a441f0a7d6f684bf8dd4d5 : SUCCESS in 1h 06m 16s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e32c9ee8067e4106b338a14ff1462679 : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/958490b6b55d472aa0a36a8d136a3acb : SUCCESS in 1h 08m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c7c2a5a67e04a3b854532a386285913 : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e515d00459d44e75b6a1efe2d862f324 : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a43f4ea540c44669f52eb72e2f490c4 : SUCCESS in 1h 24m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f5f059bc34a84e738e953381d18be665 : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc6c40b597d840f1958bbf075dd4a6b9 : SUCCESS in 1h 15m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e5ecb95036204c0ab2b7bfb19a569f99 : SUCCESS in 1h 21m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6459b293d4534463857a09b55dac8b4c : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1dbc6a528d44fa9abd90a2a944a9d0c : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2fa3ce69324445989dad911bd51848d : FAILURE in 19m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6f19645792142df924cb9cf3109776e : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e6c33ca636e4faa848776a37c81f627 : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8ad05c5e69e46fe86426d46afef7564 : SUCCESS in 1h 19m 51s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20cbd522b46d4891b10c090c357f3bd9 : SUCCESS in 1h 18m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d7ebdde6f6494314ade915baef023b5f : SUCCESS in 1h 58m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80a1517211ec44b482ce6f1b4222a469 : SUCCESS in 1h 43m 40s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/31a731a6a9994ce89af58f13aabf9ca0 : FAILURE in 1h 12m 31s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e7086e658b44bf79fe69e89d3e39c02 : SUCCESS in 1h 44m 48s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/44254874b020402c9f6e47b008cc0641 : FAILURE in 11m 00s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6ed70414f0224bb19c306864a01dcafb : SUCCESS in 1h 03m 43s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d683839dd3c43e39473fd26cf9e9d1b : SUCCESS in 1h 05m 30s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d8e37c65b2240a99f32a16e518d0de0 : SUCCESS in 1h 54m 29s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c01cd186bfed43f1bc3f3ddd83a4f91b : FAILURE in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/541935bc705c406aaf84904817824deb : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":5},{"id":"71bb7c59b408b070b93c9698f1d06191702b7aa9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-14 10:03:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2542db5032e5595be91bb1cef0d0f0ddcbe0ce7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 12:14:16.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a04a5e42d034a319d1a789f22125c67\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43054498d8454c4f86f926728991fdcc : SUCCESS in 5m 09s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdf4e4a41feb491ca7287848bdc25256 : SUCCESS in 1h 16m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/54cd1a90969940d6a3d9186b15c22431 : SUCCESS in 1h 22m 00s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c56a06ec11204e8e93e4acb097d518c5 : SUCCESS in 1h 23m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/76521868fbc24e33b3aa4354a0fbba5c : SUCCESS in 1h 05m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b7be616718e4b0e896783a7c79d191d : SUCCESS in 1h 00m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55cc4c7c8b9e4b65b42901bd06b56dec : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce586108e19d4300bf2192ec4bf7c54d : SUCCESS in 1h 17m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46126fb24e854000881a4fb6a02a89b0 : SUCCESS in 1h 04m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/233f9084bee1498b85d98fbfb9698c1e : SUCCESS in 1h 23m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e45d48611a2c4d5a8775460c7f569e6f : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/977421a1c34c4a34aa03b3ff73ff982a : SUCCESS in 45m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c15817031604470bf97539bc1befc3f : SUCCESS in 1h 15m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/08b707ea5f9b4e649a8e1672e168c9b1 : SUCCESS in 1h 04m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b9f092b991547488ae32bdf74f2a3ae : SUCCESS in 1h 11m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e5554d56ce14d089aa4ed5e8c0e8b9c : SUCCESS in 1h 27m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/96c8e408beea4475ba2f4e1b33d9056b : SUCCESS in 1h 19m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68308564519b46fa94510f5e3c1d6844 : SUCCESS in 1h 12m 33s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10ce979b4306488aa873e72451848154 : SUCCESS in 1h 16m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c847dfadc5464a57b375b6eba6437507 : FAILURE in 1h 03m 58s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2dc803a0ab54375b35d8f62c195096f : SUCCESS in 1h 10m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/441acdb8fc8148a284a4935d3206997b : SUCCESS in 2h 00m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/897f2ce1da344b65bf6b736d8262e20f : SUCCESS in 1h 43m 26s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd5973df161448ada2255dbd84b7e744 : SUCCESS in 1h 16m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2be4e01ffce5431e96755e795289b70c : SUCCESS in 1h 49m 49s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/71a905eb4876442fa50dc407c33d5c9f : SUCCESS in 1h 05m 59s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6d28567baca144c8a4a88b4753a05174 : SUCCESS in 1h 09m 17s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7083594f1044eff9734b4459de2f50e : SUCCESS in 1h 09m 21s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1cee092af56b41f5a6f3b7c13f7ab2ec : FAILURE in 29m 29s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e353b1d3647846a7b21e4bea49c1e5b4 : SUCCESS in 1h 33m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8202f6a55d804a9d9b683d7735b31aba : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":5},{"id":"ebb3995ad48f17e40f7dda186bb43e0bf74d2823","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-15 07:50:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0bb709ecd1a5e04c51eaf91af8b8bee217ea54c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 10:00:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/351450a45818419598725321ad71914e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7ec6fc5bf9d4726b6e0a1f7922f452d : SUCCESS in 6m 48s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c32461a473094e3baa2dcb1372dca612 : SUCCESS in 1h 08m 54s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be611362d36e412fa1c1000741a80499 : SUCCESS in 1h 21m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/919cc237528340d8b18ab7b9845a1a8b : SUCCESS in 1h 12m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/dea2fd92ae694a668480fb82f9dff02b : SUCCESS in 51m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50cade2bc85f4e8e9e16ee98d7e11c63 : SUCCESS in 57m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2ea64bccef840f080ec8360721428c1 : SUCCESS in 1h 09m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23e267fa92354c44bbaa5ea12890efab : SUCCESS in 1h 06m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a47e060149744efca161a985cefa56cb : SUCCESS in 51m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8523b53347f64d46a9309ddc6f3fe15a : SUCCESS in 1h 14m 53s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/027d595b22364876a161af6d0f082fe4 : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c966be5eb3bb454fa23b858029711dd6 : SUCCESS in 45m 37s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/50a9bd4cb3014cfda6cd0444382168a2 : SUCCESS in 1h 11m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/98686f2488e7423f8aae4270238afe82 : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c4a50fd56254d36ac7ee62ca2328337 : SUCCESS in 51m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39e043b3f7354e26939611aa07ee6631 : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5c7ff9d9b5a44432a25a985534587017 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d0fe983bfcc485f98ff95cea55c760f : SUCCESS in 45m 44s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11c01275e2374ff091b1082b84198d02 : SUCCESS in 1h 10m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d5b52a5f0dd41f6a2d5edcc8a08a2f3 : SUCCESS in 1h 29m 23s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9792f95b7ce9474a91ccfd822d0da7c7 : SUCCESS in 52m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a7dacf30c2a41ce9d061bf0feaebdec : SUCCESS in 1h 52m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab9364b32c114889b8bd177859ef505f : SUCCESS in 2h 01m 22s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd01c4e49130400b8f1b435ab97fbbcc : SUCCESS in 1h 55m 05s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4002abaa6f2149ebb3adbfd49077851c : SUCCESS in 1h 54m 41s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1b01a16165004510926806d409ab102f : SUCCESS in 1h 02m 27s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bb37e37d8428426682981f2d8a1b434d : SUCCESS in 1h 08m 54s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10496255ee62458e84a68a0ae2088471 : SUCCESS in 1h 02m 23s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/49b184f59adf48faa428e22078ded500 : SUCCESS in 1h 39m 20s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0093fd6a00fe4322810c614c59129bec : SUCCESS in 1h 35m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c286afcac064b9083e955297f18544c : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":5},{"id":"8826c4df7bca626b2edeb781a94a8fa8c73e1bfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-05 15:05:16.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ae50030352680b51271e93c9100cfdfbee063ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 17:17:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2253c2c70b3141ce9ed269f279486fe3\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4c16e309b4846b5bc4e9c7cbaddab92 : SUCCESS in 7m 41s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ed29eea8b6a4ac5969e20df90088720 : SUCCESS in 1h 13m 18s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b7fc1a802a7d4438aaff1d74dadc1082 : SUCCESS in 1h 22m 24s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8a73fab36a5d4c5b905cf650b94d9660 : SUCCESS in 1h 29m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a36cad3339f04ec9933a63b0f01c5218 : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29b08cac09994101b7f23309e1e339ef : SUCCESS in 1h 14m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ff16b2d98154ce28e9a5cedb9cace7c : SUCCESS in 47m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9db1df15f92c4eec8b424077d23306a2 : SUCCESS in 1h 01m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f5079746950436fb8de80d7c39dcc6a : SUCCESS in 1h 18m 39s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/428cccebdb244960a08a75866520689a : SUCCESS in 49m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3cb654facc6404e8211f63c37a7cb19 : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e52c0c0d5da74f09943a2f53e0db57bb : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/511646331ed24b048464e4ac85d7cd04 : SUCCESS in 1h 21m 49s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/33ba0eecdc11404ca07982813890c0a7 : SUCCESS in 1h 09m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/758a28783a6c498a939c6f92c310ed27 : SUCCESS in 1h 11m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b475e2f5e5be477c814f7a525b4f522f : SUCCESS in 46m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1552723077d748a6b1c4d93170e683e2 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38f79d7890674cdab2d808ca0e3931b1 : SUCCESS in 1h 20m 04s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d39acca728274d42a62959154009f7a5 : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fad4e959d6a4969ba5b4ffb47ef8353 : SUCCESS in 1h 14m 44s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f6563f5aa614b808cba4dba01c08820 : SUCCESS in 1h 22m 20s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd0e348c040e4d4a863199712eb998b2 : SUCCESS in 2h 02m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1c893cbfdcb4dc5a99aaedca6c2ccd7 : SUCCESS in 1h 52m 11s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/47ad9a4f68a24b5c827527fe738d7075 : SUCCESS in 1h 54m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85a8f1c07e2d4190b0c886bc7acbed25 : SUCCESS in 1h 55m 29s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c574e8b487b491b9e587ce43150c36c : SUCCESS in 1h 14m 23s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f7d8180e3bb247efa0a1c476000aab1a : SUCCESS in 1h 01m 57s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f3cccef9ef94b8da3ea9d3cf0f15bd3 : SUCCESS in 1h 05m 26s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b8070b84d7f741e6ae9f38bca2ada5d6 : FAILURE in 1h 34m 51s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55149f5705e24d1e9255225e21f6197c : SUCCESS in 1h 43m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c2da1968d764f688c8255fb26a98200 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":6},{"id":"06617f656ac578dd721143306a6d525328ac7c7a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-05 17:46:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a5f9bfedfd15bfcb730bce850f88c35c140aa2bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 20:32:29.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1932612fd3424cedb296328663707eba\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a04d4f50bba48049ddbdefcf7130bf1 : SUCCESS in 6m 15s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c284d5a6d458467b8b683a98b4a098c3 : SUCCESS in 1h 21m 22s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/85d07e0dd47a407cbbc3ef43bdf3e693 : SUCCESS in 1h 25m 53s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/53e1f559e31d4fc4818bdce75c7c901b : SUCCESS in 1h 30m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/87c7f9756a224d2eb5836ca54971ed4b : SUCCESS in 1h 12m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41ef260af7c84a60bbd1e2042947d318 : SUCCESS in 1h 21m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00c4819abd8e4b21a83e9b90def0fdd2 : SUCCESS in 1h 05m 13s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5eca404ca194be7b90bddabbb499f63 : SUCCESS in 1h 05m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a79a22e1f7a341f3a722e59e63eef0cf : SUCCESS in 1h 20m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/377db17c3a214c19a779d2404f4dade1 : SUCCESS in 1h 13m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00639b7cd94547b7a1a33756df633ee7 : SUCCESS in 48m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/240caf7a0d504600a5967336c6a475e4 : SUCCESS in 1h 19m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b491060fc6b4500afded22c81c50c8e : SUCCESS in 1h 15m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de38a2707a22430abfa06a13ef9abd03 : SUCCESS in 49m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea5a5aeb97e34729ab22ccb133374b66 : SUCCESS in 1h 10m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2ddfe1c49e74ddc9a0be4d53bd66787 : FAILURE in 1h 00m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad6b864828714eff83720cef71fa58c8 : SUCCESS in 1h 12m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2547f6dff7904105a23a3db44937d0f6 : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f32d7781b5a41f59aed6762e47a1575 : SUCCESS in 43m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f568a0410fdb4216a71390c75ada6d69 : SUCCESS in 1h 17m 08s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c14dfa5d6e0a455aa3f77ae8426f9aa3 : SUCCESS in 1h 11m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37a88d9485db4ccea0bd992e7ca9f06e : SUCCESS in 2h 12m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d34c366c56734a4ea0f81f0a06ba976c : SUCCESS in 1h 34m 28s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/31edcab2291743ee9df354a6b372c070 : FAILURE in 17m 44s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9faf649a3ea4794bb2fe031c0dd57db : FAILURE in 14m 22s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9a141c386c1e40249cf3d172bd2f4ae1 : SUCCESS in 45m 35s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f913697590645ba80255abf89fb40b6 : SUCCESS in 57m 27s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b96ec8e56bb4e099c24262cf936d549 : SUCCESS in 55m 39s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c4edaca94014d62a599241c13a9edf8 : FAILURE in 27m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b83e9ddc0f14dd6b4cd7f2f27956c13 : SUCCESS in 1h 26m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380b01afc4da448b949bef14f9d408ba : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":6},{"id":"eb239dc43d29bda7866a6131582ee9225258b442","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-13 07:54:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"687c2dd03803b174fbf2f4adea8a818513094e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 10:25:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c981be4c93042e9ae09f349c47c1c81\n\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f1c0be3234743e0a11b3b89882a540e : SUCCESS in 1h 17m 37s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/54a092c9ca154be2b9c1e635fa0b21b6 : SUCCESS in 1h 23m 06s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/00108c991941490baccd5ed8f868db05 : SUCCESS in 56m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/11be30970ef5446e8c507ca1f70f406d : SUCCESS in 1h 04m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dd5ab7fafb44b3e91250abcce95964b : SUCCESS in 1h 26m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceb477bff57441d9bd670a0b8e27af2b : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34795ab2c73948e7a6650548e3e3eef4 : SUCCESS in 1h 13m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d1fd6e9e22e439aae9b78e2e6da0cae : SUCCESS in 50m 56s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebf14156a79548f68d0b1b5a569aaff2 : SUCCESS in 1h 23m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2c0d1cf1d554aa0a4eb0c29400c5c08 : SUCCESS in 1h 08m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/365986ab58924868911d1acde5a3108a : SUCCESS in 1h 15m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/aae489d5db394148884d70c7ae2f456b : SUCCESS in 1h 18m 56s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3c295572c82c4810ace6ce9c9ef6da5a : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/770d0e29c23a44e6b472cba2565922c3 : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ced88afd5b57460a802ebfb5b5c6c7c0 : SUCCESS in 1h 16m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f657e04bdca946f7b46736fa1317fe17 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d97e516d47b94ae4b577df0be517bfe5 : SUCCESS in 1h 16m 32s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cb12319177a45dea05f386f3dd7dead : SUCCESS in 1h 20m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9d6d90010234bf7bc82186fb39b956f : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4f0d86964a849f1af6daad2eb3bf4fe : SUCCESS in 1h 22m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d57ee51c8aa7451ca4b6fe6bf6eede81 : SUCCESS in 2h 11m 22s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/739788aa8f484c80adf820e32e47f86e : SUCCESS in 1h 40m 21s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6aa16f6c98734e9c88e611975b0865dc : SUCCESS in 2h 04m 16s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d8bbaf6105840c8a9b0a7e8a6910366 : SUCCESS in 1h 56m 55s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/156cb3ad117d427c9ba73783f113d828 : SUCCESS in 1h 05m 51s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ed3a51751b2140b2af425e828580243c : SUCCESS in 1h 06m 57s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8894e4ed6da74885850014cc3f0219d8 : SUCCESS in 1h 00m 07s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c83053007b314aa68afc955f38c5fccd : SUCCESS in 1h 39m 18s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ae419eef17141ec85d61a9ae0438c19 : SUCCESS in 1h 17m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6477f405e4c438ebfb90e24d9f05b64 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":6},{"id":"6095e5d5db5c175b926133f2a7f4c453ee65198f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-11-18 10:23:30.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5d03113987c55df3c8283719817f459ce36b5aa8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-11-20 08:33:51.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"76ba52a4191ffcc79bdcb61b47dfa7b078499d2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 08:36:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"76bdf54cf33a4a276d67ec12c54a488071696303","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 10:36:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/294b67a1cfaa43bea6e3b8ce4ee8441f\n\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0f1a70856b894d35bcb576f9bc934730 : SUCCESS in 1h 26m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4bb6cc592141440bb8396bef6132f132 : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e1703ddbd60e4cefa6a6c8060d56959e : SUCCESS in 1h 11m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cc45945fa9a42aba1843c477ff93479 : SUCCESS in 1h 04m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb74ce03eedb464d8eaeb5dc7df56197 : SUCCESS in 1h 08m 52s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a492a6e186c34ce4a0be7cd1a5d0f175 : SUCCESS in 1h 08m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5afd6bd793c2486a87cc2a25a2a96944 : SUCCESS in 1h 09m 53s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0a1107024884eb39ee78adb47a44ec9 : SUCCESS in 1h 21m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8335390d5d447bea1e9d6241b3b55d0 : SUCCESS in 1h 18m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5b5bf3f333a147ee943499aa79aca424 : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c5b3718b2ce64497a57dccf867ea4ae3 : SUCCESS in 1h 24m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0b946faac2f45b895b5188e9565b945 : SUCCESS in 1h 14m 53s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2aa879eb460040cda27258e98e7c550d : SUCCESS in 1h 12m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/039eda9cfa7e4c91878f012d17dc1c60 : SUCCESS in 1h 24m 55s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0093b452bd8f4112b7bd566b48c682ea : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/999e6c579db94edb9269f92bf39b3efa : SUCCESS in 1h 24m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/464ea36971654f9d9321fd5fcfb15e6d : SUCCESS in 1h 00m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73c4a89987674c3e906ef289e971894f : SUCCESS in 1h 19m 50s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ec7b3b9fc6f48d89294dbef49a13186 : SUCCESS in 53m 21s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/97d366c30a334f558213c728524bf259 : SUCCESS in 1h 06m 46s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdef7ee2b12d48b1980d10f47616c753 : SUCCESS in 1h 03m 11s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/12c83341a87d486b9912460bc3995cb6 : SUCCESS in 1h 41m 40s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffec3647dd4f4ac8a942dd8b2134df13 : SUCCESS in 1h 36m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb858b7a0b440e5b9f2dcfe62815a2e : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":6},{"id":"c38606445cb65c70f33760e3603043a3b545b968","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 10:36:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"fad0cf49a89ede17d39b1e2d8fe77ca4e1527a23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 10:37:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173cd8bc9f7e4e6586a410ccb829bd8f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9de36005d54b65a59a50506b7ce088 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"78fbffd20ccec4f49756ed4a67c54887d50a8579","revisions":{"2e97f607e6e48af51e74cad3504f1cce8218fc30":{"kind":"REWORK","_number":1,"created":"2024-10-09 16:28:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/931974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/74/931974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/1"}}},"commit":{"parents":[{"commit":"d5724a0f257b944138ab6301984ae2944aab1581","subject":"Refactor rabbitmq_upgrade behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/d5724a0f257b944138ab6301984ae2944aab1581"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-09 16:28:37.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-09 16:28:37.000000000","tz":120},"subject":"Move RabbitMQ restart to handlers","message":"Move RabbitMQ restart to handlers\n\nDue to historical reasons RabbitMQ operations, like start/stop/restart\nwere perfromed in tasks and triggered on changes.\n\nAs of today there seems to be no need in such approach and we can use\nhandlers instead.\n\nIt will also move some tasks, which intend to run when service is\nrunning, to their own file. This also allows to run them once on cluster\nrather then for each host.\n\nWe also merge set_cookie with post_install as there is no reason to\nmaintain the task in it\u0027s own file.\n\nChange-Id: I222b9de3ca531724b05c7fa2cf032a3271055d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/2e97f607e6e48af51e74cad3504f1cce8218fc30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/2e97f607e6e48af51e74cad3504f1cce8218fc30"}]},"branch":"refs/heads/master"},"e7332e1fad863e83a190619001ae2a14c5a58667":{"kind":"REWORK","_number":2,"created":"2024-10-10 07:04:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/931974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/74/931974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/2"}}},"commit":{"parents":[{"commit":"d5724a0f257b944138ab6301984ae2944aab1581","subject":"Refactor rabbitmq_upgrade behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/d5724a0f257b944138ab6301984ae2944aab1581"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-09 16:28:37.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-10 07:04:11.000000000","tz":0},"subject":"Move RabbitMQ restart to handlers","message":"Move RabbitMQ restart to handlers\n\nDue to historical reasons RabbitMQ operations, like start/stop/restart\nwere perfromed in tasks and triggered on changes.\n\nAs of today there seems to be no need in such approach and we can use\nhandlers instead.\n\nIt will also move some tasks, which intend to run when service is\nrunning, to their own file. This also allows to run them once on cluster\nrather then for each host.\n\nWe also merge set_cookie with post_install as there is no reason to\nmaintain the task in it\u0027s own file.\n\nChange-Id: I222b9de3ca531724b05c7fa2cf032a3271055d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/e7332e1fad863e83a190619001ae2a14c5a58667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/e7332e1fad863e83a190619001ae2a14c5a58667"}]},"branch":"refs/heads/master"},"9bc0ba162586d15e2611ed3f0cdd16b9d3327442":{"kind":"REWORK","_number":3,"created":"2024-10-10 07:04:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/931974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/74/931974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/3"}}},"commit":{"parents":[{"commit":"d5724a0f257b944138ab6301984ae2944aab1581","subject":"Refactor rabbitmq_upgrade behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/d5724a0f257b944138ab6301984ae2944aab1581"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-09 16:28:37.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-10 07:04:48.000000000","tz":0},"subject":"Move RabbitMQ restart to handlers","message":"Move RabbitMQ restart to handlers\n\nDue to historical reasons RabbitMQ operations, like start/stop/restart\nwere perfromed in tasks and triggered on changes.\n\nAs of today there seems to be no need in such approach and we can use\nhandlers instead.\n\nIt will also move some tasks, which intend to run when service is\nrunning, to their own file. This also allows to run them once on cluster\nrather then for each host.\n\nWe also merge set_cookie with post_install as there is no reason to\nmaintain the task in it\u0027s own file.\n\nChange-Id: I222b9de3ca531724b05c7fa2cf032a3271055d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/9bc0ba162586d15e2611ed3f0cdd16b9d3327442"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/9bc0ba162586d15e2611ed3f0cdd16b9d3327442"}]},"branch":"refs/heads/master"},"4d8f64bb4d258d68e04c36cf20196b75df4918ac":{"kind":"REWORK","_number":4,"created":"2024-10-10 07:18:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/931974/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/74/931974/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/4"}}},"commit":{"parents":[{"commit":"d5724a0f257b944138ab6301984ae2944aab1581","subject":"Refactor rabbitmq_upgrade behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/d5724a0f257b944138ab6301984ae2944aab1581"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-09 16:28:37.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-10 07:18:57.000000000","tz":0},"subject":"Move RabbitMQ restart to handlers","message":"Move RabbitMQ restart to handlers\n\nDue to historical reasons RabbitMQ operations, like start/stop/restart\nwere perfromed in tasks and triggered on changes.\n\nAs of today there seems to be no need in such approach and we can use\nhandlers instead.\n\nIt will also move some tasks, which intend to run when service is\nrunning, to their own file. This also allows to run them once on cluster\nrather then for each host.\n\nWe also merge set_cookie with post_install as there is no reason to\nmaintain the task in it\u0027s own file.\n\nChange-Id: I222b9de3ca531724b05c7fa2cf032a3271055d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/4d8f64bb4d258d68e04c36cf20196b75df4918ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/4d8f64bb4d258d68e04c36cf20196b75df4918ac"}]},"branch":"refs/heads/master"},"37e73ec49121d6211086f2deb3ca6430eca61ed9":{"kind":"REWORK","_number":5,"created":"2024-10-10 14:55:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/931974/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/74/931974/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/5"}}},"commit":{"parents":[{"commit":"d5724a0f257b944138ab6301984ae2944aab1581","subject":"Refactor rabbitmq_upgrade behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/d5724a0f257b944138ab6301984ae2944aab1581"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-09 16:28:37.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-10 14:55:30.000000000","tz":0},"subject":"Move RabbitMQ restart to handlers","message":"Move RabbitMQ restart to handlers\n\nDue to historical reasons RabbitMQ operations, like start/stop/restart\nwere perfromed in tasks and triggered on changes.\n\nAs of today there seems to be no need in such approach and we can use\nhandlers instead.\n\nIt will also move some tasks, which intend to run when service is\nrunning, to their own file. This also allows to run them once on cluster\nrather then for each host.\n\nWe also merge set_cookie with post_install as there is no reason to\nmaintain the task in it\u0027s own file.\n\nChange-Id: I222b9de3ca531724b05c7fa2cf032a3271055d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/37e73ec49121d6211086f2deb3ca6430eca61ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/37e73ec49121d6211086f2deb3ca6430eca61ed9"}]},"branch":"refs/heads/master"},"78fbffd20ccec4f49756ed4a67c54887d50a8579":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-11-05 15:05:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/931974/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/74/931974/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/74/931974/6"}}},"commit":{"parents":[{"commit":"9847546fe1490abfd626a4c9a98c939a7811174a","subject":"Refactor rabbitmq_upgrade behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/9847546fe1490abfd626a4c9a98c939a7811174a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-09 16:28:37.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-11-05 15:05:16.000000000","tz":0},"subject":"Move RabbitMQ restart to handlers","message":"Move RabbitMQ restart to handlers\n\nDue to historical reasons RabbitMQ operations, like start/stop/restart\nwere perfromed in tasks and triggered on changes.\n\nAs of today there seems to be no need in such approach and we can use\nhandlers instead.\n\nIt will also move some tasks, which intend to run when service is\nrunning, to their own file. This also allows to run them once on cluster\nrather then for each host.\n\nWe also merge set_cookie with post_install as there is no reason to\nmaintain the task in it\u0027s own file.\n\nChange-Id: I222b9de3ca531724b05c7fa2cf032a3271055d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/78fbffd20ccec4f49756ed4a67c54887d50a8579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/78fbffd20ccec4f49756ed4a67c54887d50a8579"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
