)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~933724","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~master~I0d988572a102ca148059d198379d2a3115ecb795","project":"openstack/openstack-ansible-rabbitmq_server","branch":"master","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-11-26 16:24:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0d988572a102ca148059d198379d2a3115ecb795","subject":"Ensure that rabbitmq-erlang and rabbitmq-server repos use the same pins","status":"MERGED","created":"2024-10-30 10:51:23.000000000","updated":"2024-11-26 16:25:35.000000000","submitted":"2024-11-26 16:24:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933724","meta_rev_id":"e321021c0d0cda1b9e70d3e3566235f1a7408772","_number":933724,"virtual_id_number":933724,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-26 16:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-11-26 14:09:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-11-26 11:04:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-11-26 14:09:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-30 11:01:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2024-10-30 15:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-26 14:09:06.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"5e54322e5bac69dea30f6f20feed6bdb44fd6d55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-10-30 10:51:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a50a23146277547dbea45ab3841b3cc0a726df8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-30 11:01:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c26fd03e4a30b668f5a691bf736d43e8ff304bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 15:42:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c4ceaed64d140e193cd8f6213cae3a9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14433cffc4cc4eceac444fb71821ef29 : SUCCESS in 6m 59s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b662a969fd664c9f8e47d248cc08f572 : SUCCESS in 1h 20m 11s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83a1a12ce8ff4d589368270a4a23abc0 : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb40f9b6472842b6823386d577839775 : FAILURE in 29m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/baa0cd7d958343cfbc457bee40b41940 : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c883791582004f1aa1ff0df58bd5c166 : SUCCESS in 1h 10m 11s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/861cac59273443e68e3919a293ad420c : SUCCESS in 1h 13m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa33c48825fa40f5a776ed213286caa3 : SUCCESS in 1h 11m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f38d55401ee742d19a3b53a0ac9d3765 : SUCCESS in 1h 11m 39s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/088b1e099b16494292da547804338d83 : SUCCESS in 1h 12m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47bf5aea82ce4880bb9297f63b14a5d1 : SUCCESS in 1h 19m 25s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/33e5c304e75e404daa5e10a60d8cd5f9 : SUCCESS in 1h 21m 47s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5867cd7ab6e496280bab70d37250f46 : SUCCESS in 1h 20m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/150cb8b9e624467d96cf38b8ba77e722 : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b492160f49d84cd18d2e81ce33ca9044 : SUCCESS in 1h 14m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/933cacc5f6fe4e30a45951928306b21b : SUCCESS in 42m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d37a1bb5fb924491a9eebcc6084ad0c9 : SUCCESS in 1h 26m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fa5adb4107d43cd8a9a3d21293018d0 : SUCCESS in 1h 15m 11s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8532f82c03a4d828fb0cd370fb427c9 : SUCCESS in 1h 09m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ea760e845f84b07af7a75e3de842aa4 : SUCCESS in 47m 33s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0d8192132714be589d7f3db10e7066b : SUCCESS in 1h 16m 33s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b070df91b74243d19d848b11e5d1bf34 : SUCCESS in 2h 06m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23b70e05e5204645a31254d56665637b : SUCCESS in 1h 51m 41s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/429c24c9f21c4cb6bf94c8742909d8dc : SUCCESS in 2h 09m 08s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ccefbd8fba1441d9e15be9d538dbce5 : SUCCESS in 1h 43m 23s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a5923fe7fd664db1980cdc0cc92a0dfc : FAILURE in 20m 50s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2a39102b58804b959940c660b7c51921 : SUCCESS in 1h 05m 47s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/477413874ab14a79bcb1b6547b488d30 : SUCCESS in 1h 02m 56s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bfb526d8a7504625a8d197a935821a46 : FAILURE in 26m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb2f6dc0185f4d97bdc52e61b0a7ffe6 : FAILURE in 1h 02m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5f5700fb70444592120ec97df03928 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8ac18e51c2139508453c5e7ca7ae03e6fdf40917","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-04 10:52:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f9c70412a8987d9e858f6507b448552c437eea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 13:43:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45093044aa144a3683c8b96587e35390\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca3ce057189746a288991128a382cec6 : SUCCESS in 6m 47s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/09015755b35240638536b6d0eb346df5 : SUCCESS in 1h 16m 32s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c7405710c4848a087f7e258286e339d : FAILURE in 26m 26s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b2fa26fb4da7463f9c45d0ca670baa9b : SUCCESS in 1h 10m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/60bc9bba79b04ae7a3d99ddcd568e071 : SUCCESS in 57m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b68e598ce357463f98a0d438022ef2a0 : SUCCESS in 42m 31s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54e00b1905d74d28926b5e3450186d10 : SUCCESS in 51m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a77464c6ba764b6186ae9009655e5804 : SUCCESS in 45m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4be968f73cc14bd093937a82730e2fa9 : SUCCESS in 50m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39d93d9b780249beaf8ea2d18e964ac5 : SUCCESS in 1h 09m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fd7c6936a0146e69261ecf4766bd521 : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ab219059ac444f72a7cc054d4a2ab7e4 : SUCCESS in 49m 33s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e979ae0cad8447268004f49778ad8009 : SUCCESS in 1h 19m 31s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2da739c5762943cda71e6342b0b3ca21 : SUCCESS in 1h 10m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/babc1c54cf384a619555c0227a24f3b4 : SUCCESS in 46m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37d9c2477aae462ea3ec558e7df37af0 : SUCCESS in 49m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c56b0f86f02e4d57912c7253be8d5e3f : SUCCESS in 1h 26m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5ab9bf076b94332ad473bbeeaebfbe9 : SUCCESS in 1h 18m 44s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c46df176eac4bfb86bac5b34312f986 : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/695030192f3448b39783279281be3dfe : SUCCESS in 50m 43s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23aae3b153924ff2a56128305d7ed671 : SUCCESS in 47m 28s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f84cb283bb3140b0bb39a60b940d8ec7 : SUCCESS in 1h 14m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5688daf34bd747bb9ca05822ef790aee : FAILURE in 10m 23s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/47ddf88d4d7e4756ad62a716bc8cff64 : FAILURE in 18m 38s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f0acad36ea54c14bcae02e409f86e85 : SUCCESS in 1h 54m 50s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/980e4f93295345acb945a25b9adb7f5f : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fc90686bbd1747cca2fc7d04049bcc11 : SUCCESS in 1h 00m 09s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebfe5cac905b4dedb389977c7e3ce105 : SUCCESS in 1h 13m 57s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf545273cb1b4021a9b3406e982c9d0d : SUCCESS in 1h 02m 41s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6f41122fa4e4949950f8fea387ab64d : SUCCESS in 1h 22m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5346c573694fffa8fedcde837ee6ef : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bbaa1f98827d1a12bd7c2f9c103799502365172f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-04 19:03:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"843ecb28f2afd5f48f9c61974339194feb270471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 21:00:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e90f6324aef34bedb798fc005bdf151b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65c448521102441697845c537d358bbe : SUCCESS in 6m 01s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b4a14e6f23742bbb7ac27f688b8ad26 : SUCCESS in 1h 15m 11s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/065992a485e14fde81485d1014a56efd : SUCCESS in 1h 24m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dcf78dd88b274b72822de28de81961d3 : SUCCESS in 1h 26m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f2c542945cc4c6b8ebbd5d41381f640 : SUCCESS in 1h 14m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3d0c45a899d4a38933b9a1c6d5f5c39 : SUCCESS in 1h 16m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1631b6e30643493985f25783bf84c4cf : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e99d080f3e4b487982c57df14535c2a2 : SUCCESS in 1h 15m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45a80812640b44e4b7ff042153bf565d : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aff9ff9181014d44809cb6039ddd3c5c : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8641ff66b884e618522fc8f52f6fb4f : SUCCESS in 1h 26m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/253e3fc387cf48298d6929baddffee87 : SUCCESS in 1h 04m 13s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7eb58aff86424bfeb742ae12449e9179 : SUCCESS in 1h 17m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/acc2dddece2b4c9881aec68428195015 : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfe3c11819e64464995b656c878f71c4 : SUCCESS in 1h 12m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed4e83b210084026afd277871f060ad0 : SUCCESS in 1h 21m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/040fd7cfeae54e339af5be6cdc13b65d : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa786af9bcf04b6b86d907b0c127c62b : SUCCESS in 1h 17m 49s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3f613a3fdd74de6b656818f54550448 : SUCCESS in 1h 17m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0521c58c8b54e3983cb5382589a34f9 : SUCCESS in 1h 23m 42s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69f08735b5464b2dbdeb8919aef519be : SUCCESS in 46m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e98d40fa67c4b5fb991c14d29228dcf : SUCCESS in 1h 49m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/733a7ebb33324d2d8c8b3503e5af8412 : SUCCESS in 1h 49m 13s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5125ba58f3744548b8d490c62d6f35c9 : FAILURE in 1h 15m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d604dfdf08140af819ed1ee4a71b5a1 : SUCCESS in 1h 45m 01s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/361ef04636c9454da29a8760b727d21c : SUCCESS in 1h 06m 03s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5ae0b383b4054f7da378f259f4c305fb : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b01975a88f54a2f81511f39f37ac4cc : SUCCESS in 42m 14s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2813979a9ccf4953941bdd63980b9745 : SUCCESS in 1h 49m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c7c1e8cca4e4d7dbf9a646db7e8a5a5 : SUCCESS in 1h 37m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db536408312e477cbaae7f868cb349e6 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f35480543c837e6c4c62ab3ab5946d233d0e01be","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-05 09:27:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"365e21418c2cc89df29a2e397d39399fdea95bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 11:44:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41f8fe1c2c0b4fe2b2392db44a3ec53f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29830fc076b142dfad1124850031353a : SUCCESS in 7m 36s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/955832e0c92441a394d9ee1dc1146cd1 : SUCCESS in 1h 14m 26s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c69d38333b04ed7878b9343cb8d671b : SUCCESS in 1h 25m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/187a1760364149b0b996bdc461031327 : SUCCESS in 1h 19m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/005bc596b7d24bbf941951c4b7f609b1 : SUCCESS in 1h 19m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70a5625e9c734fcdbeeef069ec32b335 : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f92021d70f8441b28b772668e460f4a3 : SUCCESS in 1h 15m 22s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4eb5e31c9f84e27a628217b8453833c : SUCCESS in 1h 04m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56f89d047eda4282b3743dfe8f96f5b8 : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbc9be625f704866b956f1ae0c7e2f13 : SUCCESS in 1h 28m 51s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12047b3bdede466a874cacd18514d7a8 : SUCCESS in 1h 12m 58s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b7b25aef9c9d45738addd018fe9e9d56 : SUCCESS in 1h 13m 01s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/adca8a63336c4a77a8f10ef9f4bb66fe : SUCCESS in 1h 13m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c2cca95194374007b051dab993bad9e6 : SUCCESS in 1h 09m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1199d2b1339440f196d6cf0d24ee10ef : SUCCESS in 1h 10m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1662091aa9504f4994862e537ffcc7e7 : SUCCESS in 1h 15m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2f017c8e41f485baf890e043c42026d : SUCCESS in 1h 20m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/526234b2d473450b885eee4efb91b2c6 : SUCCESS in 1h 21m 40s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ab7fc86d798493c98327320767ff910 : SUCCESS in 1h 05m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30ca4707375041309ce559eaafc5cf4e : SUCCESS in 46m 49s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1119a88d9c2e4230b4eb21dd72807837 : SUCCESS in 1h 06m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b7372a79b924e4bb03e2ef35af7f651 : SUCCESS in 2h 00m 10s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/139f2fa8e7084fb4bf6c483fe27ad4d9 : SUCCESS in 1h 41m 19s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f63794a8947e49769048b40af135976b : FAILURE in 19m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd144148d4524a65bda6f9339ccab8f7 : SUCCESS in 1h 57m 13s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a1c79fb8796043a290c4508956e643e1 : SUCCESS in 1h 04m 43s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/334b49a7b5b34a2580a1642cea2f9db9 : SUCCESS in 1h 01m 20s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/331c37295ac44043b3e313f99dfc608c : SUCCESS in 1h 04m 21s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0cf31ebe6ecb46d7b7783574fa3541a6 : SUCCESS in 1h 52m 58s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa3aa5685c0e48f89c48de7c23b6810a : SUCCESS in 1h 37m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85fbc85c17147119c57d80e119daf65 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"70afeadfbb385d5bdca4765606e03f568c2e4abc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-05 15:23:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36e359c3df0e315607995bafbd061fc4e434fcb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 17:45:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8165a6de0f348e0a271a6c1e5746302\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7c05d670c2b4b77a26473756687fe84 : SUCCESS in 6m 34s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d6d40112bbb4840b68e795d39355031 : SUCCESS in 1h 08m 37s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d0c8ad8e551345a4877d07f6794c0cfa : FAILURE in 1h 23m 09s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b10c0a12993f42d7b272cd5b2c9d8449 : SUCCESS in 1h 15m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c007389d1dc54617a56f41ecf92fca43 : SUCCESS in 1h 13m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1cab035f916478a83ee1fecea65d031 : SUCCESS in 1h 11m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b460df7aea645b682e7e7b4af136675 : SUCCESS in 1h 11m 41s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05f7c9c2364a440f940c2d7a4113e8aa : SUCCESS in 45m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09ade06652c9453f9b00544ed0030340 : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f047f4c95928473caf5f9fa339f62257 : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b1931fc3011416b88acc5b1344a35aa : SUCCESS in 1h 15m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a21f9f0cfb06496dbd1b24359aa6de08 : SUCCESS in 1h 31m 13s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/89baba7e10b24079a0d3af6f82967bda : SUCCESS in 1h 10m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/425286d21c514a6ea1fcf689198b81ae : SUCCESS in 1h 14m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2657576b511040be909c2af435982fd9 : SUCCESS in 41m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cddc09915624990a9e9dd3241c0f658 : SUCCESS in 46m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/627ce06a16ee441897d02ed439dc3aec : SUCCESS in 1h 14m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3da101f4cee4f9bb4c97a9fd4ef9065 : SUCCESS in 1h 07m 44s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79a3c424a5f94affbf0c9afef5c12a1a : SUCCESS in 58m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee2f81025fa84dc4bfeee281dd23e1e8 : SUCCESS in 1h 13m 04s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44f8243dd6e34ca9a02454937ec9ec4f : SUCCESS in 1h 12m 14s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c54a4a3ad874de18a629c40bda635ec : SUCCESS in 1h 33m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05b0f54f9fcf4a2fb0681c239c5c41fc : SUCCESS in 2h 02m 09s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bcaa70f2f02b410cbb7089628f65287e : SUCCESS in 2h 13m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78c15ea00a914f25bd07f90b3a53b745 : SUCCESS in 1h 49m 29s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c940d8a713c44f218f41b8e7b273de6e : FAILURE in 27m 35s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6fc50d189348417b953bd626a60bbafb : SUCCESS in 1h 05m 16s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dcd60f3107e484787f1c396aa22f499 : SUCCESS in 1h 01m 11s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d56b71b71da04cfca97bc6965ed79d48 : SUCCESS in 2h 06m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58a69f644e7a4d0c96f42ffbfe4a78ce : SUCCESS in 1h 29m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f069d23c2d6d4e6682bea9bddd3926d0 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5dcc5cb58ff8989d948adf9a2464f94075d510aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-13 07:55:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_25023\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"_revision_number":2},{"id":"d313c0b61b0447326c83464a55f46ed9e8342b59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 10:18:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d42e5dbb0f4c4d0a9733d21900464e10\n\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd3280376fd74bd48c242aef7a0fe0c2 : SUCCESS in 1h 18m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/426ac980e0a543d18ede18a998d999d9 : SUCCESS in 1h 22m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d8abc17d110b4231a8c7ee03e1aa0c24 : FAILURE in 29m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fdf36b0bfa6e433aa6366cc05268ab33 : SUCCESS in 1h 08m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e711e8a7fe974a32bc49625b75cf0d4d : SUCCESS in 1h 06m 35s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29a59cbc660741f5b2a31dcd808bc1c4 : SUCCESS in 58m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f32a163a24d46febb2e702c163ad4d5 : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/735137b3284446bab664338664dbe48c : SUCCESS in 1h 20m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d62a903081944b8ae9da468ad1d3c0d : SUCCESS in 1h 15m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/394cdeb39d7d413f89451ed0e374e9e7 : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6e1e3081dd5b4994b49890e5a24d181d : SUCCESS in 1h 15m 03s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/24888bef869049d28644788d72f0d7bb : SUCCESS in 1h 22m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae4cd999c50a4a56a385a160ba54cc12 : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ce2f71b29c842bb96f75a0d6b0385ee : SUCCESS in 1h 15m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2019ff5c41e347cfa56ffb28427c35d7 : SUCCESS in 1h 14m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8aa292e1ae724b049bccc4212eeded42 : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a492cc19e744a9eac63858c86e594c8 : SUCCESS in 1h 18m 56s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa86d52c5aca4f0999deb412ff5c44b2 : SUCCESS in 1h 19m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a2a9c9874b44915b5b98738e725b14b : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4961e4e245234218b51985293802a63d : SUCCESS in 1h 22m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/50c18843f9cd48ae916caf208ce2a991 : SUCCESS in 2h 04m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4fe832c2cbd481aba4aceb081959e53 : SUCCESS in 1h 26m 59s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/74aca4fdc94c4b6696dde0e9dacc691d : SUCCESS in 2h 07m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/761378a6d14a4a26be497fa6ed782f47 : SUCCESS in 1h 38m 40s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae7cddf7294f49ada98532bfa0b98ab5 : SUCCESS in 1h 05m 55s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f3fa5d55805444d295629ae0830665d3 : SUCCESS in 1h 01m 10s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe40348caff6467cb45d7294bdb7cf2a : SUCCESS in 1h 05m 40s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/15a4d4f37a9d4647af622489a428b2e1 : SUCCESS in 1h 37m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/945d0688aad3443487f4089cdf83a118 : SUCCESS in 1h 46m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39b367117a44c46900e4dbe33f19bb3 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"05b79206df1132858b8e81a9ca7557bb44285786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-26 11:04:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_25023\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"_revision_number":3},{"id":"8cd8eceaff4e8975744bc5d490a9df22791f7053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 13:29:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/471d830dd75f44b9b118868b9f40993a\n\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/85ec1235f7e54b49b2a13d0de51e6e17 : FAILURE in 40m 24s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d7c141d3fff0472699c44a9daf164aba : SUCCESS in 1h 33m 04s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e018b2bd84a940cc80a98207593c67af : SUCCESS in 1h 28m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fa35a4c866964176a1b04aebcf28d4e7 : SUCCESS in 1h 06m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/756acfc139104aecb91b401995ebdc19 : SUCCESS in 1h 16m 46s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/048fdb44b2fe405a9eab464168f934d6 : SUCCESS in 1h 04m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d6f3986b6c4410dbc13625c6a5d5c7f : SUCCESS in 1h 05m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9875c69bac0046d3a52ec31bd524670f : SUCCESS in 1h 20m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a08f9edaee8741dfa1c9b50c39a4f07d : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa3a99edb37741438b2b1ca6b7384ecd : SUCCESS in 1h 11m 07s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/22d6f6bb7b92416d8c922d58dc440504 : SUCCESS in 1h 09m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdaf57658fa34d6d84b33f16acda5daa : FAILURE in 19m 21s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/026ec102d8aa49c6822c4cc3410a50ad : SUCCESS in 1h 20m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1335d07cf1e744d9a5097957de7de692 : SUCCESS in 1h 06m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c233b9a545054e8ab95a9f29b0bceb34 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c4bf0fbf8ad7415894a472f82c29e640 : SUCCESS in 1h 35m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6480372222cf487b85d1bb18e453c1d5 : SUCCESS in 1h 07m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecf78ae2adb945ea8a6a6602849de57f : SUCCESS in 1h 25m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b8a2f13bf424527ad09eb761a9bee13 : SUCCESS in 2h 16m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7e3e2b18d184ad89cd2592b25ea527e : SUCCESS in 1h 39m 24s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67ec04163afc47ad831b70479be20443 : SUCCESS in 1h 52m 59s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d936be96041447879c8bbd4f85f2246c : SUCCESS in 2h 08m 02s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f61277d049584eb2ac0d8afbb532582d : SUCCESS in 1h 18m 31s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f8cc1b6eb7704b68b2dc6931c69ad528 : SUCCESS in 1h 05m 06s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92872f1dadee4492a32b373dc23ee42b : SUCCESS in 1h 12m 45s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c7b73808ebff4d52a17ffd14b271abc8 : SUCCESS in 1h 59m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/226835ecc9c7450eb306e1fdce7b5e6c : SUCCESS in 1h 37m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73b3168ebc94760b654e6c74b4cae1e : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2ffcc63b5035ea9448e9580ddb5a415c02f27548","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-11-26 14:09:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1d6f3bf3cbbde4cda6d3b837384c9991d78817f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 14:11:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ee5926008fec8660700616d01cc76b7db871db53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:24:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea74bafd80434378a703c4efdb903e7b\n\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9843967d40654bcd95ae38a8cc9af5a9 : SUCCESS in 1h 26m 55s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a14629d1cdd74ccb9151d486be4d6718 : SUCCESS in 52m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0bcaca3ef1644c16aa7f2ea262ace8b8 : SUCCESS in 1h 24m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1224615567d48178ef4075b39451062 : SUCCESS in 43m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ccd69f883e744a79fb4f7aac7fb2fce : SUCCESS in 1h 20m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d13f7d4776348aea7e9c38bf5044e43 : SUCCESS in 44m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abeed0476d7d4de18fcda6a626c253e8 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5030a1e91d9c4ad6bb75591a78f6a5ea : SUCCESS in 45m 51s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d84d807e735b432ba5213c452fd640f8 : SUCCESS in 1h 21m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/69ccd015701a4c8992b17e83c8ea8d76 : SUCCESS in 1h 30m 09s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/edf3a1340c3b44108e27b66c97fcd71c : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e29eae3a8f2e45fe86b0399498e67e34 : SUCCESS in 1h 13m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ff1bdfd5571451bbad5b02ddac94c4b : SUCCESS in 1h 26m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bcc51f2bf32f4584b3b0e5a0261d4539 : SUCCESS in 1h 26m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b03aca51e944dd59419aa93f1c79bdd : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/faf10de2fed1448baecf1046d2232467 : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b2fc9899039457490ea54f3b356d8d5 : SUCCESS in 1h 07m 21s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a5c9dfbe46e8419489cd6ff09c46cb65 : SUCCESS in 1h 10m 22s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e6b12c46f13463094159c076b3fea3c : SUCCESS in 1h 09m 58s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0244db18490947a09a998acb8e1434aa : SUCCESS in 2h 04m 09s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/213180d8b2ab47828bc847a348945c65 : SUCCESS in 1h 53m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef56367fb97480494c8d3d092dc0c6b : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"36aafe66a61aaee869845d36c1857c34a8104b6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:24:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e321021c0d0cda1b9e70d3e3566235f1a7408772","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:25:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd966ed1482e40e8adb23db31e7d422d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08eded341a2a455aa1112d6260fe9b99 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0e009192879ee4f9c07d5cdc08d7071d4cd6cf2c","revisions":{"1e12ec36b4182e6f9b7bda60c561509bf87fdd65":{"kind":"REWORK","_number":1,"created":"2024-10-30 10:51:23.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/24/933724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/24/933724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/1"}}},"commit":{"parents":[{"commit":"ead47f2dbc5feae81b2febf2ef4cc91e5cd87ede","subject":"Merge \"Proceed with installation/upgrade even if cluster not healthy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/ead47f2dbc5feae81b2febf2ef4cc91e5cd87ede"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-10-30 10:47:39.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-10-30 10:47:39.000000000","tz":0},"subject":"Ensure that rabbitmq-erlang and rabbitmq-server repos use the same pins","message":"Ensure that rabbitmq-erlang and rabbitmq-server repos use the same pins\n\nThere is currently a mismatch between the apt pin priorities for the\nerlang and server rabbitmq repos. This patch sets the pinning to be\nthe same.\n\nThis should not affect normal package installation as recent changes\nto this role require installation of a specific version and will\nfail \"safe\" should that be unavailable. The intention with this patch\nis to ensure that the behaviour is the same for both the erlang\nand server repos in an abnormal situation where the installation\nmight fail.\n\nChange-Id: I0d988572a102ca148059d198379d2a3115ecb795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/1e12ec36b4182e6f9b7bda60c561509bf87fdd65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/1e12ec36b4182e6f9b7bda60c561509bf87fdd65"}]},"branch":"refs/heads/master"},"b518062de5c39f0b116490739630743a6ec12d84":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-13 07:55:12.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/933724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/24/933724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/2"}}},"commit":{"parents":[{"commit":"c7b2ef91e0d62fd0971b6bb89adb118b4d22bfae","subject":"Merge \"Move verification of cluster_state to a separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/c7b2ef91e0d62fd0971b6bb89adb118b4d22bfae"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-10-30 10:47:39.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-11-13 07:55:12.000000000","tz":0},"subject":"Ensure that rabbitmq-erlang and rabbitmq-server repos use the same pins","message":"Ensure that rabbitmq-erlang and rabbitmq-server repos use the same pins\n\nThere is currently a mismatch between the apt pin priorities for the\nerlang and server rabbitmq repos. This patch sets the pinning to be\nthe same.\n\nThis should not affect normal package installation as recent changes\nto this role require installation of a specific version and will\nfail \"safe\" should that be unavailable. The intention with this patch\nis to ensure that the behaviour is the same for both the erlang\nand server repos in an abnormal situation where the installation\nmight fail.\n\nChange-Id: I0d988572a102ca148059d198379d2a3115ecb795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/b518062de5c39f0b116490739630743a6ec12d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/b518062de5c39f0b116490739630743a6ec12d84"}]},"branch":"refs/heads/master","description":"Rebase"},"0e009192879ee4f9c07d5cdc08d7071d4cd6cf2c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-26 11:04:18.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/933724/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/24/933724/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/24/933724/3"}}},"commit":{"parents":[{"commit":"795d141c851f20ac509be4c1c1d2e8e776f732ee","subject":"Merge \"Respect `package_state` for the role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/795d141c851f20ac509be4c1c1d2e8e776f732ee"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-10-30 10:47:39.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-11-26 11:04:18.000000000","tz":0},"subject":"Ensure that rabbitmq-erlang and rabbitmq-server repos use the same pins","message":"Ensure that rabbitmq-erlang and rabbitmq-server repos use the same pins\n\nThere is currently a mismatch between the apt pin priorities for the\nerlang and server rabbitmq repos. This patch sets the pinning to be\nthe same.\n\nThis should not affect normal package installation as recent changes\nto this role require installation of a specific version and will\nfail \"safe\" should that be unavailable. The intention with this patch\nis to ensure that the behaviour is the same for both the erlang\nand server repos in an abnormal situation where the installation\nmight fail.\n\nChange-Id: I0d988572a102ca148059d198379d2a3115ecb795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/0e009192879ee4f9c07d5cdc08d7071d4cd6cf2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/0e009192879ee4f9c07d5cdc08d7071d4cd6cf2c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
