)]}'
{"id":"openstack%2Fopenstack-ansible-rabbitmq_server~946234","triplet_id":"openstack%2Fopenstack-ansible-rabbitmq_server~master~I92f24c77b5fb8a3dd9ed17915d7c725d15f9f21e","project":"openstack/openstack-ansible-rabbitmq_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-06 07:50:13.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I92f24c77b5fb8a3dd9ed17915d7c725d15f9f21e","subject":"Ensure that failures are fatal for upgrade_check","status":"MERGED","created":"2025-04-03 10:16:58.000000000","updated":"2025-05-06 09:52:23.000000000","submitted":"2025-05-06 09:51:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946234","meta_rev_id":"43b127f233aca032a2ceb9d3355ae04a79876739","_number":946234,"virtual_id_number":946234,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-06 09:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":2,"date":"2025-05-05 13:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-14 07:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":1,"date":"2025-05-06 07:50:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-03 12:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-14 07:52:17.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-04-18 01:28:40.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"CC"},{"updated":"2025-05-05 13:45:14.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"0651147a564f8f415192f823f3bc135c9613c76a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-03 10:16:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"602c405b4e8f9596b9dcd381ca76294d9a65b405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 12:21:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/143b55cb5627460391ab9206a32f61dc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a79456db79c417595ec8d1f6d2eecf5 : SUCCESS in 6m 53s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0b0233105064d54a16073e00eb93b8c : SUCCESS in 10m 21s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/17f4d9fc8009436981e5092772161836 : SUCCESS in 4m 59s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/00075a9f99f445c4a76556065cee6120 : SUCCESS in 4m 40s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c3bb79b67f94348b1d825bd36ed4dd8 : SUCCESS in 5m 26s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3ca5f7069b748e1aa2dd2b97767d1a5 : SUCCESS in 8m 19s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d3d2f9803f744df9fac5cca5d6a85e8 : SUCCESS in 1h 05m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d63ba220777342f2aee2297f8aaa0156 : SUCCESS in 1h 12m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/42bc5bf943264b00af7e3284606db4ed : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a4efc86f88624fdf92bc4c8d0d4ca00a : SUCCESS in 1h 01m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5cc1474f6684b97a542b7a55248c0cb : SUCCESS in 1h 04m 11s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d207ceb77cd43b29dd73ad5dc49e1ef : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f146654186c49378a1f8699a43b027f : SUCCESS in 1h 13m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c89992ef3da34a73a99bb503be871cd2 : SUCCESS in 1h 12m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd534b3078bd4a349a8741ccab2b4a22 : SUCCESS in 1h 20m 53s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f510d1da02564783b87963a4b25c2320 : SUCCESS in 42m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/996bf7fb0a3e426bbf881b96fecdfe35 : SUCCESS in 1h 12m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d89ad6f8f3274ce3947fcd920a01f07a : SUCCESS in 38m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dca3a760992049da91befee57dae10ca : SUCCESS in 1h 15m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c11545f8d4240b791b1b6d76e2069d3 : SUCCESS in 1h 13m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fab2e1deef0b496ca442b5ed260a9d00 : SUCCESS in 1h 26m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b13815fdf5e44d4a8e53e6f61d62e5d0 : SUCCESS in 1h 18m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/077d39f460a44c7585068f2c30feb857 : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5498bf74cb8047c9aec4c8adba3a5c13 : SUCCESS in 1h 22m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/906eb5117d6744ed833f79766618ddd3 : FAILURE in 1h 58m 43s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/518dae34594140629b834b4a952fe2be : SUCCESS in 1h 46m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8c1db13f4704f82bcf165cb60c560c2 : SUCCESS in 1h 54m 48s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47cbe309f1124effb885ea55d8a1c325 : SUCCESS in 1h 39m 31s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f3767882733842aabd45006ed98fe189 : SUCCESS in 1h 56m 08s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5097ad3296864ccb8759db7e5d0eccb5 : SUCCESS in 51m 18s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2f40cd0353ff48daaa0a995a6f5ee0a8 : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/03cbd979c9f44841846609d35a8538f7 : SUCCESS in 1h 08m 24s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c5f03504d504110b2099f51632d01b9 : SUCCESS in 57m 49s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/53da730bf8ee4d41a3cc305ad46b29fd : SUCCESS in 1h 43m 24s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37e4c4de0de446eb84934d656da9b9ea : SUCCESS in 2h 00m 54s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84c31a1294da4783983056f513e4dfa3 : SUCCESS in 1h 22m 38s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6178027a8334b67802a0cf6283be455 : SUCCESS in 1h 27m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e25c5f6f9e2d472b93ba9ca3c4fdbed1 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"84c12866100e28d0f4abf7dd28114b5aae9cc3ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-03 15:11:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7d777bb1e5122068f90ad6e4edb2f2e4d4b8951","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-03 15:36:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ebb9e8d6948dc8d78d56576b7ba7f01096763d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 17:24:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7d5f305e597410ebd444cc434962b74\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1d50acb58354519ab913860b3101442 : SUCCESS in 8m 53s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/853bf68e7ecc4936a68820222a5ce1ce : SUCCESS in 8m 58s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/c54cc7b28c404949965e7b4a2e1becfb : SUCCESS in 11m 26s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b24a4d5c308d455eaf8b10b9ff3d1e68 : SUCCESS in 10m 53s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb72b427a5e7495e87464434b9aecac2 : SUCCESS in 5m 39s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a08cee3d1f24fbe8ba3d3523a1249a0 : SUCCESS in 8m 01s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1886670ab17c4fd0b668f96f6cde0b1b : FAILURE in 39m 24s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9e794752e8b24b13ad3bc21cbb663fff : FAILURE in 35m 02s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/361a3fadef1043f79261f253c506d78f : FAILURE in 39m 56s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8369a837e821486880e5c96597f55313 : SUCCESS in 1h 11m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e22bea8199fa4f5cb488e823a61ee813 : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/372bf82ecd1148348567320a11424942 : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/456e3f2c40834c65b8f30d311a5412db : SUCCESS in 1h 10m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e572a5566c5246a293778698bb42141d : SUCCESS in 1h 07m 05s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d4ad0841986410eb4427dbb62c4845c : SUCCESS in 1h 39m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8fdb64d472948fca51ad9a7fc5e9f72 : SUCCESS in 1h 13m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2108a54742fe4556bdead5428819a04d : SUCCESS in 1h 17m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c201a35b3344708ab0d9871df012175 : FAILURE in 19m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c0bbe3003ee46af8b41b285ec5682ca : SUCCESS in 37m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e672ca666960418493cf59c471a3755b : SUCCESS in 1h 05m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e86bdf5858324a9088979065599af655 : SUCCESS in 1h 17m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/868543021343426c989cb61f32c42eec : SUCCESS in 45m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20beede106d14ba781f71be0dda08892 : SUCCESS in 38m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2d11e88528847ce857af8d0da991d2e : SUCCESS in 1h 18m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c777ea56bb04f3fa2dd78b8e977551b : FAILURE in 18m 14s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/03ff51d708814b7fa296e3d31e4cd45f : FAILURE in 1h 15m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/506cc0ef9a154ff1a91c06fadb8c9cb9 : FAILURE in 13m 51s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46d97fb3366b46fbaa7d4469cd5766db : SUCCESS in 2h 05m 18s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8519b895dd2f44bfa22f235d9a59983b : FAILURE in 17m 20s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a4f4df5e2604b2eaba1b7b680b34d44 : SUCCESS in 1h 51m 58s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ccf2ca65889344c1a13f3c02be107765 : SUCCESS in 1h 12m 06s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/00c2f6ae5c67417dabbca3d6f12de124 : SUCCESS in 58m 44s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee40e7b7d25f4d5aa4107ef760ea919a : SUCCESS in 1h 14m 06s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f7c9100bbdf741e9bc37a34ceaa9b926 : FAILURE in 30m 11s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7fe17338f8274b2f8c3544e73a6cbc41 : FAILURE in 1h 25m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02fed0b1fc86429a8ce069b1a334bda5 : SUCCESS in 1h 39m 03s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/930d092f301c4a3cb58451f2c0266165 : SUCCESS in 1h 32m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd0821ff2cbd4ed59fc8f788053a8608 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f969d216293fca1d2c8fce912b43fd0e0f44d659","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-04 07:51:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e49a07a9b2eb64b7b1ec4dde7b4aea96e09fb06c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 10:14:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b02b7b04aeb841dc95fe957b83f73ff9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc9da22d647343d6a67b9af411ef0050 : SUCCESS in 7m 21s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/91628c47e3d745bb803b48e29930d5b7 : SUCCESS in 8m 44s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/ea565b5b62934ad98ad5cad6dfeaf2be : SUCCESS in 10m 09s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5406376f04974fc79eed83e9eb02a8fa : SUCCESS in 8m 52s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e150f1681894ad1bae7c9f41f2f874d : SUCCESS in 9m 38s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/795bde747e184ea7bf5f4dcefa73601e : SUCCESS in 9m 42s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/58f99fa356c64309a05bf9ef0f8fc205 : SUCCESS in 1h 13m 52s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa557c10a26b437dbf45be7625c7465c : SUCCESS in 1h 15m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/13c56085be6347bf9a3c4c3f8af061fd : SUCCESS in 1h 08m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b31070ad249e4ea9b33d8d4c4f9a9140 : SUCCESS in 48m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0d1047185394c2b940efae9f23fcd09 : SUCCESS in 56m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6a7ad6060244a6593737be7749db15a : SUCCESS in 1h 05m 13s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63ed460901b940ae9f03f8ba2269f0dc : SUCCESS in 1h 07m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbca0440bb594d44bac4bb168b59c202 : SUCCESS in 1h 09m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cc9c16925ca40f8b39c91989e1c7e48 : SUCCESS in 1h 17m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb4c9fc290e549f4a8f550444a083c36 : SUCCESS in 1h 14m 57s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/035dfaf300e14ea4883320f683530e8e : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0589a34886df4e18b7212503acf440cb : SUCCESS in 1h 21m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/461e82bd0fd544939c0fcc554abcb8cb : FAILURE in 52m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ac40d7e4df54e23a67dabd043a1f015 : SUCCESS in 1h 09m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/896d74efc4d84d84a7856956a6e2a785 : SUCCESS in 1h 19m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51af641cbd4d4ebba41fa9dd7420de55 : FAILURE in 53m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35bed13bcb8b437e80e3d28c25207d67 : SUCCESS in 1h 13m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b371303df9f7447db9bf2071226c0c25 : SUCCESS in 1h 21m 23s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b3e4363f55a14c6ba079f1768c099221 : FAILURE in 54m 49s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/95b257ada7d248669be7ca5e89c819fb : FAILURE in 1h 38m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/323a536e83c44d5c9e96dc8582c27822 : SUCCESS in 1h 55m 06s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e89d8b79e05646ee95f80bd342b1c335 : SUCCESS in 1h 51m 41s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/52d049547de34a3b8c9b6dcb64de7bc9 : FAILURE in 1h 21m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d89507461a9741b4aa25daa28746af0b : SUCCESS in 2h 05m 11s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bfad3f28ad0b4f05b0448b029a569f7d : SUCCESS in 1h 06m 36s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f925df78d78c478f8fef8eb2a3799db3 : SUCCESS in 59m 11s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3a76be4b5064a7ca0c727dfb39d3ccb : SUCCESS in 1h 13m 17s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8a61fc345508453d861b989f086aa009 : FAILURE in 1h 48m 01s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a15ba35d0254dac95770af4bd6b22d4 : SUCCESS in 1h 40m 25s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddc1d8b16d994a4abb5357e745809dd0 : SUCCESS in 1h 33m 56s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7473557c6d0402da9e6e309158bb3b0 : SUCCESS in 1h 31m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b17744c47947418b84009c704da701 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":1},{"id":"25d5cfa94eefb0535ed1e0412e7a48598a59feee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-10 13:15:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"892259e20b07f4a77aed3e18337beb4ed00e2cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 15:23:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a232c3f9cf02448187916d35ff7172a2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0062fea9e93f4549840fff31962baec3 : SUCCESS in 5m 53s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e472e21ee23496b8f483da13d3bf20c : SUCCESS in 8m 22s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/3ef165168a7341ae800f881cd825eb50 : SUCCESS in 9m 03s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1514379cc75e4d818d8cb2898c5230ba : SUCCESS in 10m 27s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/333f467cdc8f4760824d7d857a173024 : SUCCESS in 9m 01s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58bac07a44a54672a67aa67fd0ba7f11 : SUCCESS in 5m 08s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9235b42dea74b3bb6690936b34bda90 : SUCCESS in 1h 16m 18s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/36e60292bf6d4cfd92fbed80baf76a5f : SUCCESS in 1h 24m 23s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/931918179407431dbdbb85bc969cb6e3 : SUCCESS in 1h 14m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bdc1f3bcb3ec43c4901722dd0cfabc0b : SUCCESS in 1h 23m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf4d40dc1476462a88336f383654ca01 : SUCCESS in 1h 15m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f85a13360c76484b801f71091529881b : SUCCESS in 1h 03m 47s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15868951b0324c96a0f180ad8229257b : SUCCESS in 1h 12m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/705923394bcd400085f6f575ed338d96 : SUCCESS in 1h 09m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49ffcf2d189f499bafe04e2a78303a44 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d64de854d304d2f83d9b2e3040698c6 : SUCCESS in 1h 10m 43s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/89adad8d47c6444cb872a3115e0fe504 : SUCCESS in 1h 14m 18s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/090c6c8a32dd48a4a475556fcde83b80 : SUCCESS in 1h 20m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/925ade4ce2354a3b8adb76596ff7176b : SUCCESS in 1h 18m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85ac88f98b674eeeb01ba4746cfa4158 : SUCCESS in 1h 06m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/beb117e082cc41dda4c6fb0199568270 : SUCCESS in 1h 38m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7e8aca3b1f834de7b5c6ec4ee0c79759 : SUCCESS in 1h 12m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71476f429f7b4a8684c7685e4bf5da0a : SUCCESS in 1h 08m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/902fc1bd203d4dad8bfb77c97d97d407 : SUCCESS in 1h 16m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/13f86540b6794dab901ea2bbb715b2f0 : SUCCESS in 2h 04m 36s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/88010818ed62495c919359ff5442b9c1 : SUCCESS in 1h 51m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/121a1b2ac7bc416aac87b1f5c477acf0 : FAILURE in 48m 58s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c20a1d066d084841b74126a69f57fbee : SUCCESS in 1h 55m 39s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c0bbe113f3d40d9b779509a686934e3 : SUCCESS in 1h 49m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cb9dcae230b41c3a34f067dc434b174 : SUCCESS in 1h 47m 20s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a1c4c4f3a6f0478fae74e9279eec8cec : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/66fe1647355a4a86b3323685c981143e : SUCCESS in 1h 03m 00s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2766a2934128471c80ce2d82a5329dc7 : FAILURE in 57m 08s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/090005e1b7154d75b7dc437ec4bfc3da : SUCCESS in 1h 44m 53s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/443d787a0ac744ba838f1c8e36392ab1 : SUCCESS in 1h 43m 38s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0330bab7b494f97998b5bd65dc4313c : SUCCESS in 1h 32m 55s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bb12808366d4b00b5a6127dad1e64a0 : SUCCESS in 1h 28m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/201fd7046d26400dbbcb8d6705f03391 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4d2051fd6176f9571712c5def36bfba68a862aea","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-04-14 07:52:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2c70bbbd6a998e3b6bd10323a8c78bb27903f332","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-04-14 07:52:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"deff19fcdec98d66774906c14b9c8c9495f111c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 10:08:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a33082d4296941969b0d6e0f330a610f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71437de8684c49758a7f436cf874ae2f : SUCCESS in 5m 28s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4694cd8431a4daa8a80d87bc8eb907c : SUCCESS in 10m 25s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/66ffc8a3b18a483db762b5ae06778fe1 : SUCCESS in 11m 59s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7276a48c82ff489daa74b7c9d893f0f7 : SUCCESS in 9m 05s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a3c9b07264c487daabfdbd562f12250 : FAILURE in 4m 45s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3e391ae06a24f14b53aa45f96e224df : SUCCESS in 10m 10s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3ad18b6be7447cd8b81578a9a75eabf : SUCCESS in 1h 09m 22s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f578e540450b4f4f8d607a7ef3507547 : SUCCESS in 1h 19m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d86b8ad308224971ab6ed54035fbcf5b : SUCCESS in 1h 18m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/aca849c8ca4f47bcb8c6959c522e6b8c : SUCCESS in 1h 05m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8f5fda4f6714e1ea53824f69389e3b1 : SUCCESS in 1h 12m 20s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2d243e0c172400195d4a9a92cf624ea : SUCCESS in 1h 06m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b29737a973f4c5c8f6e13d40f23c65e : SUCCESS in 1h 10m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90aca719f74149f7a28b8a974fa2370c : SUCCESS in 1h 17m 46s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca605a3dde764ebfb1b5bdb9ef81c98b : SUCCESS in 1h 17m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70a650b9e52e441ca19092f0cce0e5f4 : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2af3a65507794542a3bd29e95b4d3f43 : SUCCESS in 37m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d4ac572c4334462a5912adc077d030b : SUCCESS in 37m 56s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f2e2ff21fa524ea7a792a2c340b4a0e5 : SUCCESS in 1h 10m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c84b54470354f3aaed2d5d16a1c8938 : SUCCESS in 1h 06m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b4bb270983048adbb3c9a56345e9ad6 : SUCCESS in 1h 02m 37s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c70cdfd5cadb42f6937cd2e33faeaa09 : SUCCESS in 1h 21m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bb07c2cc78f4987ba57c2ddf8853fa2 : SUCCESS in 1h 15m 10s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e17e9bff1de843b0a6111027d2789f39 : SUCCESS in 1h 25m 05s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d89984d198b4bb1921f55d5a6103a68 : SUCCESS in 54m 16s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5fe511f122dd46d8b2a981f2d4760122 : SUCCESS in 1h 51m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1ec8923db984d9f955664fc652b4e54 : SUCCESS in 2h 02m 01s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b48fff6a818c41e4be3c5116e844eb11 : SUCCESS in 1h 55m 10s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f7e40d7efc164c70ac7afe7c2f5168c1 : SUCCESS in 1h 55m 55s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df54dbc8fd6046268ee790b98c848b2a : SUCCESS in 1h 59m 08s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3b735c02e2834b179006cc93f06475cb : SUCCESS in 1h 08m 02s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/52736a10405a4019b320f4aca285b3c0 : SUCCESS in 1h 05m 15s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1f1b8d5b59047cc92b19c737de08ce8 : SUCCESS in 1h 09m 56s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3beb957d803e4091b28a08b0eda09fc2 : SUCCESS in 1h 58m 24s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c154045ef9449898c2f257f1e86785a : SUCCESS in 1h 49m 56s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5361613ef8d4569bf97e4f9ca27bb35 : SUCCESS in 1h 33m 45s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/047e1b4144f142568399f96d9c7fd094 : SUCCESS in 1h 30m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c4eab18e6f48148c826ea2b02ef743 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"07c5535e0fa90d3e19d1462afa81907112c0a008","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2025-04-18 01:28:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7ad9f5b6d4376ee7bb3c5318cfd6732fe0676d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 03:36:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3a99a5a880c49a19a0a1e671de8e0ba\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bba838be2fb54f07be38f8a5a34fb362 : SUCCESS in 4m 48s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab8391abca9f49d5a487a6a0beb984f1 : SUCCESS in 6m 06s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/52d68e675d814f63a97686ab98f60672 : SUCCESS in 6m 05s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9219d838076b444da7f11292d29bf35e : SUCCESS in 5m 58s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b389f479138840159f939de9204eed12 : SUCCESS in 8m 42s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25dd2beeb94842bab992f8adfab20cf3 : SUCCESS in 10m 30s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2d487ccf7874510b646ad915c85bc9e : SUCCESS in 1h 06m 06s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d4e48a504e7944a29d6a182729116dd9 : FAILURE in 38m 58s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/60c22d0187574b86b02b4008c4f2aade : FAILURE in 49m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0231a7e03afa4c028a1b39debad5786d : FAILURE in 1h 03m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37e734770820411fb0d9eead6152b684 : SUCCESS in 1h 06m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbfadef3fa19456ead4a633c12fbddcb : SUCCESS in 1h 13m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ae95fd4f9a846499fc62659de38f31f : SUCCESS in 1h 10m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cd16bccb6f54b38ab6a4006262c0e50 : SUCCESS in 1h 14m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61b3548d6d714b86bed13819d900ff4b : SUCCESS in 1h 12m 22s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0e62b9757954e759340d8b75a246dad : SUCCESS in 1h 19m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9cdd1d4be4804b31a0c4626a42393f9f : SUCCESS in 1h 18m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c01e9105ba64df883bfbd3ce2cfe1af : SUCCESS in 1h 15m 56s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f5857f70a3a437ca42e9815309ef121 : SUCCESS in 1h 20m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bd55babd2c14bd0ae9a349abc18a031 : SUCCESS in 1h 23m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b42a1ed77d9e4272a4e6b102bcd2d6a8 : SUCCESS in 38m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd71e3a3e5dc40c689f8bca66aa7fc2b : SUCCESS in 1h 24m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddbe6a28036f45cdaa31ae545b980f53 : SUCCESS in 38m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d406a3e483344b8ad033bb9b1fbfb8d : SUCCESS in 38m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/467e23c7416b4772b96912211db10e7f : SUCCESS in 2h 00m 27s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1862b93d74904f7097961e5bb9a1b37d : SUCCESS in 55m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/706d6c5edb01441fbd191d9942c88d88 : SUCCESS in 53m 54s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dca0209ecf504af9bff903438584df35 : SUCCESS in 57m 46s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a25f51663c9542998707f950089539ed : SUCCESS in 1h 02m 48s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ae46e6899b04af981fd4c6bde604c29 : SUCCESS in 1h 03m 23s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/337096a9edae4811a038f3179da7cfa3 : SUCCESS in 39m 48s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ac00392f4d3e4b79a149b21e7c79f51b : SUCCESS in 1h 02m 00s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdcfc212aa964600839a01c0bb512756 : SUCCESS in 57m 31s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/23dcc3337c904fe295dee428adfd4429 : FAILURE in 44m 07s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b60c7dcfb3d4fa8be8151ca58e97d89 : SUCCESS in 1h 33m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46813adb34124690ba5af59a71c34883 : SUCCESS in 1h 25m 58s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e98543cc25846289382bcbc8106ac9d : SUCCESS in 1h 19m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90cba33111e84a76bbf763b9c45671d5 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"723429a112aafbfe8fab5248825cfbf1ca711786","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-25 14:53:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee2578263742dfbc7a9564d433738db7ee4b8df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 17:09:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d96a692688ea4621b5d638f1ef541a55\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87060a20b766464e918a69f02217c2d8 : SUCCESS in 6m 20s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/0307b0619a7443d4a6f5f89331af54ab : SUCCESS in 8m 31s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/f499f83a1b1e4394bbe29357999952d3 : SUCCESS in 5m 03s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/089bb2f6431a4eef878894fdf6b18ce4 : SUCCESS in 4m 55s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4e9ab1083874ed79a68132dca1878bc : SUCCESS in 5m 08s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08c56878218343b4b5704b8d141c67bf : SUCCESS in 9m 58s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a20cc8520c96476dbc2ef9d7a1d4deb2 : SUCCESS in 40m 55s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d8c690cdb6f43b2aaa7d86fc269f3ef : SUCCESS in 1h 13m 31s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/711816ba4a6b4bf985e5e415cd600e45 : SUCCESS in 1h 16m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7d8ba461b11445c3a12b4eb618a4c085 : SUCCESS in 1h 19m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c06ebfe6896b4302813ceeba6c0e545f : SUCCESS in 1h 08m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/146517132274480f8ae32cb6a8fa5480 : SUCCESS in 1h 15m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3514722205b44299287b0962d3d3ac8 : SUCCESS in 1h 04m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e338ef72729a4a4b8dbd8b2e817c249a : SUCCESS in 1h 08m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/603c9a351238406388cf56f646f6b164 : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70ad1f47765f4e92b62d8477cc062db2 : SUCCESS in 1h 06m 49s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/33601189b0fe4089b5c9c466ebc339d3 : SUCCESS in 1h 01m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d53425b28296490596bb919213f8309f : SUCCESS in 1h 07m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b5b7787d3e345689d978fc69036b095 : SUCCESS in 1h 11m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3209c49c39d94f77894e0b182299098e : SUCCESS in 59m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d1097e3e23d4a1ea41b0ea553e28131 : SUCCESS in 1h 06m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/310a5703e7ab43a1af78e7c10e6f7a36 : SUCCESS in 47m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2790afacadd04bb8a511b51ed63749b6 : SUCCESS in 1h 17m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9115e37e6a9b4f53885049cec4ec3866 : SUCCESS in 1h 22m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4c39b28a9f87439e8978722bda7c9c5f : SUCCESS in 2h 03m 34s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a86024aca518468c8948b80efddf9249 : SUCCESS in 1h 59m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49b4521d2d4f4bf58a75319f49a58fd8 : SUCCESS in 1h 54m 32s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61329069d1cc42cda56010910772fd33 : SUCCESS in 1h 55m 27s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c606fbfdc7d4539a4abe2d455f116b4 : SUCCESS in 1h 58m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca8861b1e2c241b78abe6ab3eb57795f : SUCCESS in 1h 40m 50s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aeef460059bf4718a49127c5bce09aa3 : SUCCESS in 1h 07m 36s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f589937727946bd8721f1151a6385d8 : SUCCESS in 1h 06m 27s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dc7253eef0346ca8c8f20cc2f49d12d : SUCCESS in 38m 17s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1eafd2060891457d8ae95935ac7a3596 : SUCCESS in 52m 53s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/645c366a688443629cee4034c4fe5167 : SUCCESS in 56m 04s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87353b817e174d22843092d2342a9314 : SUCCESS in 1h 27m 31s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b29afa88395548dd95d3f02469d32126 : SUCCESS in 49m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa60ada1ac554a7abeb718998acf33f7 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"087952cba1a7a83171f01d201ee8480640b02987","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2025-05-05 13:45:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f7ef86da028c7c890bb4702c31e1ad0033b84993","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-06 07:50:13.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7f7bd861ea03f38fe575326f83434f546ecec65d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:52:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c56edbed8458e6ace6574c35c09c88a91e4be68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 09:51:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d8f99701f8640b79fde0dfd00446365\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d1639176caf4255a41575216f8990ef : SUCCESS in 6m 09s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a245089f0814c79925751c64b4bb56f : SUCCESS in 8m 30s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/6dcdcd1c71ba406f92f2e2672bdee635 : SUCCESS in 8m 44s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0797669fd0024ef788ee16a3754a9d85 : SUCCESS in 9m 07s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f62b328e94e49b3b381a1b5bedc787f : SUCCESS in 7m 33s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c21a582117442a4bd33eda84adbcf9a : SUCCESS in 11m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e1fdeec496b646cbbb6a81854990195b : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3a3be93851204946b8ca4818b33c32b8 : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/decee08af7494ab19bce2be1c175aa3c : SUCCESS in 1h 11m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe4e56ff59d547b9a8566fa26ad910d6 : SUCCESS in 1h 10m 15s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ab18b17cfdb41288f6fbea389d14e09 : SUCCESS in 1h 11m 10s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/339cddf9e16e49378f09096a8d94b8f3 : SUCCESS in 1h 11m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbeca8ad9e744d73923b2cf29cdd778e : SUCCESS in 1h 28m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df423326761a4afc8c0f91e710e3b4ad : SUCCESS in 1h 08m 22s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09a8fc888f6447f18d479d4c302f4126 : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9b1a227727be4aacb9898446dbed0aa0 : SUCCESS in 1h 07m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b3f3f1b8fb414f9c8af991dcbfae84b6 : SUCCESS in 1h 21m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1d2728d9c1d4244b8146039f754feef : SUCCESS in 1h 10m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d28f793dd281485ca603edef0d0841c3 : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6d7d9b3aa5e74876ac8d8fad35f9a8c2 : SUCCESS in 1h 22m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/351f057b7dbc40acaf58c934eb383c99 : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/836ebdf4f15f4cae8033b57b33099310 : SUCCESS in 1h 26m 52s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39761d6dfc29446eb218b6c92f88ee28 : SUCCESS in 1h 03m 07s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f73818ef84ba4a9dbd85a32eb13f929c : SUCCESS in 1h 11m 15s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9970a6abef2484c9adf2fe6f1a97ecd : SUCCESS in 1h 07m 33s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a215ff824c6f4f4da58c470226be0721 : SUCCESS in 1h 45m 37s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/10154fe7fe584bd691a69b0d9017f808 : SUCCESS in 1h 54m 51s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df0e39ae42a64bafb190aeb1deb5e125 : SUCCESS in 1h 40m 34s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5401df827f444199bfa632de150e7b9 : SUCCESS in 1h 40m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593f366cbdc24a40921c98be38afe563 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"ff62a3562edbe732583eed2794a7b599beed2ca1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 09:51:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"43b127f233aca032a2ceb9d3355ae04a79876739","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 09:52:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/321a0e16668c44d5a023244d67bf095f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02f44feb976a4ae78c0191f25755e0b9 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"485ee87076cd2df1585a80b52e3e167943b0e009","revisions":{"19b689905aee004aa28aff9701069b601742f1d3":{"kind":"REWORK","_number":1,"created":"2025-04-03 10:16:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/34/946234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/34/946234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/34/946234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/34/946234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/34/946234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/34/946234/1"}}},"commit":{"parents":[{"commit":"cdabf2d3e8126ec89951f608d020282c57a40247","subject":"Fail with human-readable errors if upgrade impossible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/cdabf2d3e8126ec89951f608d020282c57a40247"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-04-03 10:16:48.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-04-03 10:16:48.000000000","tz":120},"subject":"Ensure that failures are fatal for upgrade_check","message":"Ensure that failures are fatal for upgrade_check\n\nAnsible\u0027s fail will fail only current host in case of met conditions\nwhile continuing play for others.\n\nIn order to prevent further execution for all hosts in case of unmet\nconditions, which may further break the cluster, `any_errors_fatal`\nwas added to fail tasks.\n\nThis will end the play for current batch and cancel further batches.\n\nChange-Id: I92f24c77b5fb8a3dd9ed17915d7c725d15f9f21e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/19b689905aee004aa28aff9701069b601742f1d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/19b689905aee004aa28aff9701069b601742f1d3"}]},"branch":"refs/heads/master"},"485ee87076cd2df1585a80b52e3e167943b0e009":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-10 13:15:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/34/946234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server","ref":"refs/changes/34/946234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/34/946234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/34/946234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/34/946234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rabbitmq_server refs/changes/34/946234/2"}}},"commit":{"parents":[{"commit":"4655c9108974efa647882607e2c064c4ec981a39","subject":"Fail with human-readable errors if upgrade impossible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/4655c9108974efa647882607e2c064c4ec981a39"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-04-03 10:16:48.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-04-10 13:15:10.000000000","tz":0},"subject":"Ensure that failures are fatal for upgrade_check","message":"Ensure that failures are fatal for upgrade_check\n\nAnsible\u0027s fail will fail only current host in case of met conditions\nwhile continuing play for others.\n\nIn order to prevent further execution for all hosts in case of unmet\nconditions, which may further break the cluster, `any_errors_fatal`\nwas added to fail tasks.\n\nThis will end the play for current batch and cancel further batches.\n\nChange-Id: I92f24c77b5fb8a3dd9ed17915d7c725d15f9f21e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/485ee87076cd2df1585a80b52e3e167943b0e009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rabbitmq_server/commit/485ee87076cd2df1585a80b52e3e167943b0e009"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
