)]}'
{"id":"openstack%2Fopenstack-ansible-repo_build~690321","triplet_id":"openstack%2Fopenstack-ansible-repo_build~stable%2Ftrain~I1eae97c73528266bd5435244e5d56dc284453d2f","project":"openstack/openstack-ansible-repo_build","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"I1eae97c73528266bd5435244e5d56dc284453d2f","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","status":"ABANDONED","created":"2019-10-22 19:04:02.000000000","updated":"2020-04-05 17:59:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6097ba0ecfb4efcd617cfecb16228662b2b63661","_number":690321,"virtual_id_number":690321,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-22 19:37:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-11-18 10:32:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 19:37:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-18 10:32:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"8706864aaf4070fb6d66f651ae54010371f7b5ea","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-10-22 19:04:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5be2d0c795d1d11e29be105edcdc0ead21743d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:37:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/23bf81dd0a50453da6407aa6ea31339d : SUCCESS in 6m 15s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/ac34b34d44b543fe933988fadd4415a2 : FAILURE in 14m 18s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/03abc2e2af2e43a8b580e3ac20a29445 : FAILURE in 15m 44s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5dc5ad61d6143db8663b693d5e59b56 : SUCCESS in 16m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9df5a7b0f3c545d89a0e0e1fceea0463 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aef4fb4f41db48b6b97e640bef031560 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"da8015fe4f2fd85b6ac0d865b6ebb504da761dfa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-24 10:38:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce4b284087bd076d465c0f1fab5765b972c86a19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 11:11:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6f6d79cb146e404c9be150adca75f3e5 : SUCCESS in 9m 13s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/320d8b1c51284488b31f45cfb62d8e22 : FAILURE in 10m 59s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/39506404afaa4b61a7d0458ced08aa79 : FAILURE in 14m 33s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7fe58b7553934c618767e57915424c48 : SUCCESS in 25m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc067a5f40143878b2d3a706890bdaa : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6ffd4a529054882a716d37f9fc7b0e5 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"37fef2016032ec2d55511010600bd5ea0b627716","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 10:32:18.000000000","message":"Patch Set 1: Code-Review-1\n\nWonder if we should be retiring the repo_build repo now as it\u0027s not needed in master or T.","accounts_in_message":[],"_revision_number":1},{"id":"84f2d9517c52e7018ca61a39da13b84b269f8947","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-05 17:59:01.000000000","message":"Abandoned\n\nretiring role","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35469c6cd9d21e435e02b26ff957825894f2a126","revisions":{"35469c6cd9d21e435e02b26ff957825894f2a126":{"kind":"REWORK","_number":1,"created":"2019-10-22 19:04:02.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/21/690321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_build","ref":"refs/changes/21/690321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_build refs/changes/21/690321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_build refs/changes/21/690321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_build refs/changes/21/690321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_build refs/changes/21/690321/1"}}},"commit":{"parents":[{"commit":"08c7e8321fad52501bbaf1999b184da1f91f677d","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_build/commit/08c7e8321fad52501bbaf1999b184da1f91f677d"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-22 19:04:00.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-22 19:04:00.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/train branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I1eae97c73528266bd5435244e5d56dc284453d2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_build/commit/35469c6cd9d21e435e02b26ff957825894f2a126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_build/commit/35469c6cd9d21e435e02b26ff957825894f2a126"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
