)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~614136","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I17260a844d42be232e9976bd409a303fc5212882","project":"openstack/openstack-ansible-repo_server","branch":"master","hashtags":[],"change_id":"I17260a844d42be232e9976bd409a303fc5212882","subject":"missing document start \"---\"","status":"MERGED","created":"2018-10-30 08:04:58.000000000","updated":"2019-04-24 20:04:05.000000000","submitted":"2019-04-24 20:04:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"614136-1556136245968-9fef03c1","meta_rev_id":"b7635419a0992cffb6758f037cfabf19d4c63b6f","_number":614136,"virtual_id_number":614136,"owner":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-24 20:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-04-24 18:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-04-24 18:35:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 18:35:22.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 20:04:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"980cdda821cdb6134ab7069d891cf5864ee6e272","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-10-30 08:04:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"605361e57372199622caa23693e07dd490e8bbc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-30 08:35:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/36/614136/1/check/openstack-ansible-linters/12d67ae/ : SUCCESS in 6m 26s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/36/614136/1/check/openstack-ansible-functional-centos-7/c6f91f4/ : SUCCESS in 21m 28s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/36/614136/1/check/openstack-ansible-functional-opensuse-150/9d5756f/ : SUCCESS in 17m 09s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/36/614136/1/check/openstack-ansible-functional-opensuse-423/257e59a/ : SUCCESS in 18m 09s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/36/614136/1/check/openstack-ansible-functional-ubuntu-bionic/81c8506/ : SUCCESS in 21m 35s\n- openstack-tox-docs http://logs.openstack.org/36/614136/1/check/openstack-tox-docs/5a00354/html/ : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a28315d5c672e01c5e1af9ff4ce806336dac0a35","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 18:35:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62de209263f8b3224b3d02303e1c524ba35d040c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:35:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8db866b222a301df77795d0781373a7ad7cafa6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 20:04:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b7635419a0992cffb6758f037cfabf19d4c63b6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 20:04:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/36/614136/1/gate/openstack-ansible-linters/93c06f3/ : SUCCESS in 5m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/36/614136/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/a6efb65/ : SUCCESS in 1h 00m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/36/614136/1/gate/openstack-ansible-deploy-aio_metal-centos-7/a146ed7/ : SUCCESS in 1h 18m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/36/614136/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/c6983c3/ : SUCCESS in 1h 11m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/36/614136/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d810977/ : SUCCESS in 59m 41s\n- openstack-tox-docs http://logs.openstack.org/36/614136/1/gate/openstack-tox-docs/4d4e342/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ce809363c1a40c5e31e8f5f17982cc094e9b187","revisions":{"4ce809363c1a40c5e31e8f5f17982cc094e9b187":{"kind":"REWORK","_number":1,"created":"2018-10-30 08:04:58.000000000","uploader":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"ref":"refs/changes/36/614136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/36/614136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/614136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/614136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/614136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/614136/1"}}},"commit":{"parents":[{"commit":"467f8e30805301645956f1e3d72a933cf47495b1","subject":"Use distro packages only after they are installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/467f8e30805301645956f1e3d72a933cf47495b1"}]}],"author":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-10-30 08:04:33.000000000","tz":0},"committer":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-10-30 08:04:56.000000000","tz":0},"subject":"missing document start \"---\"","message":"missing document start \"---\"\n\nChange-Id: I17260a844d42be232e9976bd409a303fc5212882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4ce809363c1a40c5e31e8f5f17982cc094e9b187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4ce809363c1a40c5e31e8f5f17982cc094e9b187"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
