)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~657309","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I2b76a7ed792ee9c17816cce5f46b84eb588aedd6","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"openstack-ansible-repo_server","hashtags":[],"change_id":"I2b76a7ed792ee9c17816cce5f46b84eb588aedd6","subject":"Change the tests source to opendev","status":"MERGED","created":"2019-05-06 03:19:42.000000000","updated":"2019-06-20 15:35:37.000000000","submitted":"2019-06-20 15:35:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657309-1561044937988-83facb33","meta_rev_id":"ce82d01e646bd9c4decc1adf2c2bcdf977f27a97","_number":657309,"virtual_id_number":657309,"owner":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"date":"2019-06-20 12:21:51.000000000","_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-20 15:35:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"value":2,"date":"2019-05-21 10:15:37.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-21 11:07:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-21 11:07:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-08 15:13:43.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-21 10:15:37.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-21 11:07:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-20 15:35:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a05cabc867e8482488313f7efc30fcec73e540e9","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-05-06 03:19:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43990fa86f71c898cae244cf1467c5004abe5b49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 04:37:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/657309/1/check/openstack-ansible-linters/d97d033/ : SUCCESS in 4m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/09/657309/1/check/openstack-ansible-deploy-aio_metal-debian-stable/76667da/ : SUCCESS in 54m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/09/657309/1/check/openstack-ansible-deploy-aio_metal-centos-7/53c3a83/ : SUCCESS in 1h 12m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/09/657309/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/02268af/ : SUCCESS in 53m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/09/657309/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/649c9f2/ : SUCCESS in 52m 46s\n- openstack-tox-docs http://logs.openstack.org/09/657309/1/check/openstack-tox-docs/3689bcd/html/ : SUCCESS in 2m 33s\n- build-openstack-releasenotes http://logs.openstack.org/09/657309/1/check/build-openstack-releasenotes/8c41bb1/html/ : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b65d9f91ec8bfbb5af8111f5fa59f1ebd37d1962","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-08 15:13:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"825630072e3a69a1d4c38b85707c05def35eb3b8","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-05-21 08:53:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ba01bf3c966ee03ffc0cfc3b37aa02fccfa67eb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:11:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/657309/2/check/openstack-ansible-linters/8be2597/ : SUCCESS in 6m 59s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/09/657309/2/check/openstack-ansible-deploy-aio_metal-debian-stable/8b33332/ : SUCCESS in 1h 01m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/09/657309/2/check/openstack-ansible-deploy-aio_metal-centos-7/79b1f90/ : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/09/657309/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/d709c8e/ : SUCCESS in 1h 06m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/09/657309/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c6b472d/ : SUCCESS in 53m 48s\n- openstack-tox-docs http://logs.openstack.org/09/657309/2/check/openstack-tox-docs/b366c9b/html/ : SUCCESS in 2m 58s\n- build-openstack-releasenotes http://logs.openstack.org/09/657309/2/check/build-openstack-releasenotes/038576f/html/ : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4591f3bf12b587afe8ab519552c10dc3a9aa389c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 10:15:37.000000000","message":"Patch Set 2: Code-Review+2\n\nWow, with the good published url.","accounts_in_message":[],"_revision_number":2},{"id":"0f0ee9279fec4e745fe893e9c7c898e472a656fa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-21 11:07:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c76dc332c1de8002f40556541fa1a2776e57a2e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 11:07:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e75e805bf893098c8507bf673fcf4d96bef70603","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 12:31:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/09/657309/2/gate/openstack-ansible-linters/8ffd4b6/ : SUCCESS in 5m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/09/657309/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/51c7268/ : FAILURE in 52m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/09/657309/2/gate/openstack-ansible-deploy-aio_metal-centos-7/63c00f2/ : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/09/657309/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/ee6dc62/ : SUCCESS in 1h 05m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/09/657309/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b75c7b7/ : SUCCESS in 56m 40s\n- openstack-tox-docs http://logs.openstack.org/09/657309/2/gate/openstack-tox-docs/4308b09/html/ : SUCCESS in 2m 37s\n- build-openstack-releasenotes http://logs.openstack.org/09/657309/2/gate/build-openstack-releasenotes/7237618/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4eb947063f7e009306a0a1ded8df9f94804a0dcf","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-06-20 12:21:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c15079b07950b7e8af8414047cc881933ef147a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 14:01:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/657309/2/check/openstack-ansible-linters/5b0cbb9/ : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/09/657309/2/check/openstack-ansible-deploy-aio_metal-debian-stable/d42650a/ : SUCCESS in 1h 25m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/09/657309/2/check/openstack-ansible-deploy-aio_metal-centos-7/f7a06ad/ : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/09/657309/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/7907264/ : SUCCESS in 1h 07m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/09/657309/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2b37db1/ : SUCCESS in 59m 30s\n- openstack-tox-docs http://logs.openstack.org/09/657309/2/check/openstack-tox-docs/0f82784/html/ : SUCCESS in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/09/657309/2/check/build-openstack-releasenotes/0282d32/html/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"d30e2d10e2d069fff4bca454a17ebc605d0c1c12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 14:01:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cd553cc08a26c5870167de050c64d482233854d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 15:35:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/657309/2/gate/openstack-ansible-linters/db4c30e/ : SUCCESS in 4m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/09/657309/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/27d32f8/ : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/09/657309/2/gate/openstack-ansible-deploy-aio_metal-centos-7/3a25a50/ : SUCCESS in 1h 28m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/09/657309/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/6ca8cc8/ : SUCCESS in 1h 16m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/09/657309/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/dba39a9/ : SUCCESS in 57m 12s\n- openstack-tox-docs http://logs.openstack.org/09/657309/2/gate/openstack-tox-docs/706b5b6/html/ : SUCCESS in 2m 48s\n- build-openstack-releasenotes http://logs.openstack.org/09/657309/2/gate/build-openstack-releasenotes/ae64fbe/html/ : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":2},{"id":"6147e5a3b8bbddde0473098199dc49e73ac587cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 15:35:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9136da9015ec8e98b71228e956b5a82bd3033f2f","revisions":{"5c637b1b8a01c8fa245da54de45d91443aad9fe8":{"kind":"REWORK","_number":1,"created":"2019-05-06 03:19:42.000000000","uploader":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"ref":"refs/changes/09/657309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/09/657309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/09/657309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/09/657309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/09/657309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/09/657309/1"}}},"commit":{"parents":[{"commit":"0b4eab0089f9cdd2ea698f2b5ee3235a39f4674a","subject":"Merge \"missing document start \"---\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0b4eab0089f9cdd2ea698f2b5ee3235a39f4674a"}]}],"author":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-05-06 03:16:43.000000000","tz":480},"committer":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-05-06 03:19:39.000000000","tz":480},"subject":"Change the tests source to opendev","message":"Change the tests source to opendev\n\nChange-Id: I2b76a7ed792ee9c17816cce5f46b84eb588aedd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5c637b1b8a01c8fa245da54de45d91443aad9fe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5c637b1b8a01c8fa245da54de45d91443aad9fe8"}]},"branch":"refs/heads/master"},"9136da9015ec8e98b71228e956b5a82bd3033f2f":{"kind":"REWORK","_number":2,"created":"2019-05-21 08:53:17.000000000","uploader":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"ref":"refs/changes/09/657309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/09/657309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/09/657309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/09/657309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/09/657309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/09/657309/2"}}},"commit":{"parents":[{"commit":"0b4eab0089f9cdd2ea698f2b5ee3235a39f4674a","subject":"Merge \"missing document start \"---\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0b4eab0089f9cdd2ea698f2b5ee3235a39f4674a"}]}],"author":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-05-06 03:16:43.000000000","tz":480},"committer":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-05-21 08:53:09.000000000","tz":0},"subject":"Change the tests source to opendev","message":"Change the tests source to opendev\n\nChange-Id: I2b76a7ed792ee9c17816cce5f46b84eb588aedd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9136da9015ec8e98b71228e956b5a82bd3033f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9136da9015ec8e98b71228e956b5a82bd3033f2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
