)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~657350","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~Ia67cc43f2f10abafb0426695adeec73bf771c1a7","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"update_tests","hashtags":[],"change_id":"Ia67cc43f2f10abafb0426695adeec73bf771c1a7","subject":"Updated the git source","status":"MERGED","created":"2019-05-06 08:51:15.000000000","updated":"2020-10-29 00:14:21.000000000","submitted":"2020-10-29 00:07:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657350-1603930049707-06587f2b","meta_rev_id":"61893c50a1c97c744f36771db5829c3b435afd85","_number":657350,"virtual_id_number":657350,"owner":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"date":"2020-10-21 03:27:21.000000000","_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-29 00:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"value":2,"date":"2020-10-28 17:45:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-21 11:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"value":1,"date":"2020-10-28 17:45:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-08 14:52:30.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-10-21 11:32:51.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-28 17:45:46.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-29 00:07:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa32e64ca3c9e67e0e1bf24a40f8a89ce2011c20","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-05-06 08:51:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"466b36dd6b9aadc226c89cc5a3c143e6a63335c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 10:07:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/50/657350/1/check/openstack-ansible-linters/0636b5f/ : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/50/657350/1/check/openstack-ansible-deploy-aio_metal-debian-stable/415269d/ : SUCCESS in 1h 04m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/50/657350/1/check/openstack-ansible-deploy-aio_metal-centos-7/991569f/ : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/50/657350/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/18ed9e1/ : SUCCESS in 1h 04m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/50/657350/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/26dd712/ : SUCCESS in 1h 06m 03s\n- openstack-tox-docs http://logs.openstack.org/50/657350/1/check/openstack-tox-docs/240e232/html/ : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e7a9acafffb18da95c0dde05174e14b38f8bfe45","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-08 14:52:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56f5317d4b26299908a67f3e7f230a7e4e932b20","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-05-13 03:14:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e8375821393e2a663b63a6bc5d39ca19af0b06fc","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-05-13 03:14:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2903b03adf06583d1c3eb91d35678ed4c70aaac0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 04:42:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/50/657350/2/check/openstack-ansible-linters/2b5411d/ : SUCCESS in 5m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/50/657350/2/check/openstack-ansible-deploy-aio_metal-debian-stable/f82a138/ : SUCCESS in 52m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/50/657350/2/check/openstack-ansible-deploy-aio_metal-centos-7/c411f4d/ : SUCCESS in 1h 15m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/50/657350/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/e2050b0/ : SUCCESS in 55m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/50/657350/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/fc25426/ : SUCCESS in 46m 13s\n- openstack-tox-docs http://logs.openstack.org/50/657350/2/check/openstack-tox-docs/6459498/html/ : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":2},{"id":"61a4513cc533f8656e6e55c3b4ce8de29378fa57","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-06-20 12:23:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"df2b709eb09c597d108e29478fff4991f9d45f1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 13:53:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/50/657350/2/check/openstack-ansible-linters/a1dfede/ : SUCCESS in 4m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/50/657350/2/check/openstack-ansible-deploy-aio_metal-debian-stable/491f7f0/ : SUCCESS in 1h 05m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/50/657350/2/check/openstack-ansible-deploy-aio_metal-centos-7/4653ba8/ : SUCCESS in 1h 23m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/50/657350/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/f944f3c/ : SUCCESS in 1h 07m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/50/657350/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d2a75a9/ : FAILURE in 33m 09s\n- openstack-tox-docs http://logs.openstack.org/50/657350/2/check/openstack-tox-docs/09b7891/html/ : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"9de29be523cdde4e271df7844da7b4caff7c8ad8","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2020-10-21 03:27:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ea798ab16404d66e4666c365443271e7826c85d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 05:48:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bb724881459d4217ae7af87667fbe189 : SUCCESS in 9m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/18ffaad47db34897969358faec51af7d : SUCCESS in 1h 13m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7333c69f3e4d433e89fa3ff14a4c4594 : SUCCESS in 1h 29m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/725ae4cd62274ff5b1fcf816bc9f0ec6 : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94f1303bf9f94aeca7a05b700d83b90a : SUCCESS in 1h 10m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0aa0f433e4774eb2ac0af4649798087e : SUCCESS in 2h 13m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c32cff49daa84cc18dd4b03ea18fc489 : SUCCESS in 1h 30m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c623a30fc794c558575e4e10454d161 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":2},{"id":"7c2641628b638c629b706cdee9a327a17f6a26b5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-21 11:32:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"74ec52365d90ec7de9a10c47afe7ef7ba33cc2f7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-28 17:45:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"06d4a913b644c53a0b312c7e473bc5ecf008fe12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 17:45:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"436fdb394ad3af3ecf6c934bf45cc932d273b908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 19:47:21.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0bc195c213554c1ca90a3106cba2c357 : RETRY_LIMIT in 4m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f26aaeb8983a46f6a2418cc2dc4e6245 : SUCCESS in 1h 03m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/043701b28d124193b682c3d4cdff7fa4 : RETRY_LIMIT in 59m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4945a1aea6f44a43a7d7351f522cd8b5 : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ecf561819ad422fa413da43022d3883 : SUCCESS in 1h 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb4532dc558c41c1b8160f74bda54462 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d576f1866b39800a17de32ab47b75c86a64f0145","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-28 20:17:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6a763c305bead49d301383a8d9bf9292b9f9885a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 22:38:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/152b462f155a4b72921a9fec15fc55d1 : SUCCESS in 9m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6a9b805d1f7348388572d47c3cd6d6a7 : SUCCESS in 1h 13m 56s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7719a331ea044afeadd0d1378fc3e8b3 : SUCCESS in 1h 27m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93b40c29b3714a41a79cc0a9add49dda : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04fb4b560a0c47cbb7b005fad8d59819 : SUCCESS in 1h 04m 52s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/37c90b3a22834c26b1233ac84d3b0309 : SUCCESS in 1h 57m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bf9ba04e5704a488e9367a748f3c034 : SUCCESS in 1h 54m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe1b90b6386541199465e80e900471e7 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"3cf2852c7ac9e0ed4f4ba2c2341dd068e1a31d42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 22:38:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"785dce95dc419691ab9a4e7cafe99e4cfeb321ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 00:07:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d4bc15b351744b7d98517cd2dc07cfb3 : SUCCESS in 9m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bba09a1cf9c947e5a3392bd1303eeab9 : SUCCESS in 1h 10m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fb4439a3ba424fbea370b30695aabbe4 : SUCCESS in 1h 24m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5c7da20934d4a789e1a50b27f26cf99 : SUCCESS in 1h 14m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc8d6881435f4954b29774b5c8ab6e6a : SUCCESS in 1h 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d899fffcf1d9465c800c86b0ee80eac2 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6c9f440017ed2409b6f357f1798f24a26e9723eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 00:07:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a067a3bdbf2035908ef2c877bd741a7f142805dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 00:14:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96001b67fc8148b58cd47239d8da0dc1 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0713fcf2b3162e2c65a7ed631eaaa6f36b9fa472","revisions":{"b3ea8c547c9b56a3363995aabc0a86460c6bef48":{"kind":"REWORK","_number":1,"created":"2019-05-06 08:51:15.000000000","uploader":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"ref":"refs/changes/50/657350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/50/657350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/50/657350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/50/657350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/50/657350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/50/657350/1"}}},"commit":{"parents":[{"commit":"0b4eab0089f9cdd2ea698f2b5ee3235a39f4674a","subject":"Merge \"missing document start \"---\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0b4eab0089f9cdd2ea698f2b5ee3235a39f4674a"}]}],"author":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-05-06 08:50:58.000000000","tz":480},"committer":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-05-06 08:50:58.000000000","tz":480},"subject":"Updated the git source","message":"Updated the git source\n\nChange-Id: Ia67cc43f2f10abafb0426695adeec73bf771c1a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b3ea8c547c9b56a3363995aabc0a86460c6bef48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b3ea8c547c9b56a3363995aabc0a86460c6bef48"}]},"branch":"refs/heads/master"},"0713fcf2b3162e2c65a7ed631eaaa6f36b9fa472":{"kind":"REWORK","_number":2,"created":"2019-05-13 03:14:01.000000000","uploader":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"ref":"refs/changes/50/657350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/50/657350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/50/657350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/50/657350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/50/657350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/50/657350/2"}}},"commit":{"parents":[{"commit":"0b4eab0089f9cdd2ea698f2b5ee3235a39f4674a","subject":"Merge \"missing document start \"---\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0b4eab0089f9cdd2ea698f2b5ee3235a39f4674a"}]}],"author":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-05-06 08:50:58.000000000","tz":480},"committer":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-05-13 03:13:41.000000000","tz":0},"subject":"Updated the git source","message":"Updated the git source\n\nChange-Id: Ia67cc43f2f10abafb0426695adeec73bf771c1a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0713fcf2b3162e2c65a7ed631eaaa6f36b9fa472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0713fcf2b3162e2c65a7ed631eaaa6f36b9fa472"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
