)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~658048","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~Ia58f4b417fb6e6f8ddf6f53427a7aba2abbeffab","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ia58f4b417fb6e6f8ddf6f53427a7aba2abbeffab","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:36:05.000000000","updated":"2019-05-09 15:00:19.000000000","submitted":"2019-05-09 15:00:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658048-1557414019919-e62a4e3b","meta_rev_id":"d90a12db08cad973bf8c94944f8eaf3c42482bb3","_number":658048,"virtual_id_number":658048,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-05-09 15:00:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-05-09 12:10:03.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-05-09 12:10:03.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:10:03.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-05-09 15:00:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"781732bc0c048d9b0b2c1f8ea250b27bee7898b0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:36:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25c7b314b99eeae897bcca94aee5895d52d63960","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-09 12:10:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68623d7d0a817729fbff0f41577cab1019a1b690","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:22:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/48/658048/1/check/openstack-ansible-linters/76cc7a7/ : SUCCESS in 4m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/48/658048/1/check/openstack-ansible-deploy-aio_metal-debian-stable/d366cf4/ : SUCCESS in 56m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/48/658048/1/check/openstack-ansible-deploy-aio_metal-centos-7/2198a62/ : SUCCESS in 1h 21m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/48/658048/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/cefd999/ : SUCCESS in 1h 07m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/48/658048/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/917fc98/ : SUCCESS in 52m 09s\n- openstack-tox-docs http://logs.openstack.org/48/658048/1/check/openstack-tox-docs/30b2687/html/ : SUCCESS in 2m 33s\n- build-openstack-releasenotes http://logs.openstack.org/48/658048/1/check/build-openstack-releasenotes/5a45202/html/ : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3191c8cf6d1491d2e4f93bca6112017c581cd2c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:22:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6010a0c19cca1f3c44fd26d3108c3a06c8439943","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:00:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f5a2cff0f11206a359fdf7ffb54d407ff43297b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:00:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/48/658048/1/gate/openstack-ansible-linters/5ecd80d/ : SUCCESS in 4m 52s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/48/658048/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/d79cd3f/ : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/48/658048/1/gate/openstack-ansible-deploy-aio_metal-centos-7/f4cfe79/ : SUCCESS in 1h 31m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/48/658048/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/850fff5/ : SUCCESS in 1h 07m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/48/658048/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c777bba/ : SUCCESS in 55m 38s\n- openstack-tox-docs http://logs.openstack.org/48/658048/1/gate/openstack-tox-docs/e67b31a/html/ : SUCCESS in 2m 10s\n- build-openstack-releasenotes http://logs.openstack.org/48/658048/1/gate/build-openstack-releasenotes/bce5a65/html/ : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"582a189117e861f8ceb73e8fe1e1d20dda083c59","revisions":{"582a189117e861f8ceb73e8fe1e1d20dda083c59":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:36:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/658048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/48/658048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/48/658048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/48/658048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/48/658048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/48/658048/1"}}},"commit":{"parents":[{"commit":"0b4eab0089f9cdd2ea698f2b5ee3235a39f4674a","subject":"Merge \"missing document start \"---\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0b4eab0089f9cdd2ea698f2b5ee3235a39f4674a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:36:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:36:04.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ia58f4b417fb6e6f8ddf6f53427a7aba2abbeffab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/582a189117e861f8ceb73e8fe1e1d20dda083c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/582a189117e861f8ceb73e8fe1e1d20dda083c59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
