)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~658251","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~Id1e8cc503d71cf6e7d0d84b5a74f6c3d7bce9393","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa/el8-support","hashtags":[],"change_id":"Id1e8cc503d71cf6e7d0d84b5a74f6c3d7bce9393","subject":"[redhat] Install rsync-daemon for newer releases","status":"MERGED","created":"2019-05-10 03:49:19.000000000","updated":"2019-05-10 19:51:12.000000000","submitted":"2019-05-10 19:51:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658251-1557517872229-cdafc7c2","meta_rev_id":"99d9d993d2822de9aa851821e1bbe3fd1b38a878","_number":658251,"virtual_id_number":658251,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-10 19:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-05-10 18:33:05.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-05-10 17:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-05-10 18:33:05.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 17:33:20.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-10 18:33:05.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-05-10 19:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e2e36fc58702de4603497b65a5d52fc4fedeedd","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-10 03:49:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1afcab4c5ef6616d52716f640bbc8dfb016a5312","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-10 04:25:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"457f63fdd68ce87f9129d26e2a5219b980faf437","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 06:35:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/658251/2/check/openstack-ansible-linters/d471d1d/ : SUCCESS in 4m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/51/658251/2/check/openstack-ansible-deploy-aio_metal-debian-stable/7452f0a/ : SUCCESS in 59m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/51/658251/2/check/openstack-ansible-deploy-aio_metal-centos-7/6f6356d/ : FAILURE in 14m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/51/658251/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/d826e90/ : SUCCESS in 1h 04m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/51/658251/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6798025/ : SUCCESS in 50m 55s\n- openstack-tox-docs http://logs.openstack.org/51/658251/2/check/openstack-tox-docs/85376f3/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"55d1837da52a5ca75c0e4678fa883a5d71ddd93b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-10 16:26:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1a620da73102b78fd02335a672cef61719699242","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:48:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/658251/3/check/openstack-ansible-linters/e4f01d0/ : SUCCESS in 7m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/51/658251/3/check/openstack-ansible-deploy-aio_metal-debian-stable/e2d904c/ : FAILURE in 10m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/51/658251/3/check/openstack-ansible-deploy-aio_metal-centos-7/b377a9f/ : FAILURE in 13m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/51/658251/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/5bc78f4/ : FAILURE in 14m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/51/658251/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/1c31b65/ : FAILURE in 11m 59s\n- openstack-tox-docs http://logs.openstack.org/51/658251/3/check/openstack-tox-docs/7ba0380/html/ : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":3},{"id":"d1c658dd9b9da51651f736513c68c7fa3655a143","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-10 16:56:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6eb4b4a9994e86658ca457c49a636888591fa4de","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-10 17:33:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9ccabef1b10078299899e272ee05f8fb0e6d2f18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:11:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/51/658251/4/check/openstack-ansible-linters/4eece19/ : SUCCESS in 4m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/51/658251/4/check/openstack-ansible-deploy-aio_metal-debian-stable/4c49d9b/ : SUCCESS in 1h 02m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/51/658251/4/check/openstack-ansible-deploy-aio_metal-centos-7/85884da/ : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/51/658251/4/check/openstack-ansible-deploy-aio_metal-opensuse-150/16b68d4/ : SUCCESS in 1h 05m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/51/658251/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c690ab2/ : SUCCESS in 59m 57s\n- openstack-tox-docs http://logs.openstack.org/51/658251/4/check/openstack-tox-docs/29507e0/html/ : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":4},{"id":"2dacca22722c7009574ec4e8d5847dcd0fd13e0c","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 18:33:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8a756896b42b9efba3b52f6f4fff9dddf5567d08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:33:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5926a12e00b6f841f477d46041ee9334cb3d6e9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:51:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/51/658251/4/gate/openstack-ansible-linters/8c558ec/ : SUCCESS in 8m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/51/658251/4/gate/openstack-ansible-deploy-aio_metal-debian-stable/a9d8946/ : SUCCESS in 1h 01m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/51/658251/4/gate/openstack-ansible-deploy-aio_metal-centos-7/b5293e7/ : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/51/658251/4/gate/openstack-ansible-deploy-aio_metal-opensuse-150/6324316/ : SUCCESS in 1h 06m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/51/658251/4/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a844f10/ : SUCCESS in 55m 36s\n- openstack-tox-docs http://logs.openstack.org/51/658251/4/gate/openstack-tox-docs/0c4be71/html/ : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":4},{"id":"8806cdc1003a4c370598d5e636bdb1238d3012d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:51:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ba7b358ccd9dbe042ee6cdec0706abfecd7ce631","revisions":{"42439e760eace54e0913c1958d8c4205a6039b0e":{"kind":"REWORK","_number":1,"created":"2019-05-10 03:49:19.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/51/658251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/51/658251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/1"}}},"commit":{"parents":[{"commit":"582a189117e861f8ceb73e8fe1e1d20dda083c59","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/582a189117e861f8ceb73e8fe1e1d20dda083c59"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 03:48:18.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 03:48:18.000000000","tz":-240},"subject":"[redhat] Install rsync-daemon for newer releases","message":"[redhat] Install rsync-daemon for newer releases\n\nThe newer packages havee moved rsyncd into its own package called\nrsync-daemon so we\u0027ll install that for anything newer than EL7.\n\nChange-Id: Id1e8cc503d71cf6e7d0d84b5a74f6c3d7bce9393\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/42439e760eace54e0913c1958d8c4205a6039b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/42439e760eace54e0913c1958d8c4205a6039b0e"}]},"branch":"refs/heads/master"},"0b789378e684003ad64e22874d77a4c6f160b9d0":{"kind":"REWORK","_number":2,"created":"2019-05-10 04:25:38.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/51/658251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/51/658251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/2"}}},"commit":{"parents":[{"commit":"582a189117e861f8ceb73e8fe1e1d20dda083c59","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/582a189117e861f8ceb73e8fe1e1d20dda083c59"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 03:48:18.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 04:25:32.000000000","tz":-240},"subject":"[redhat] Install rsync-daemon for newer releases","message":"[redhat] Install rsync-daemon for newer releases\n\nThe newer packages havee moved rsyncd into its own package called\nrsync-daemon so we\u0027ll install that for anything newer than EL7.\n\nChange-Id: Id1e8cc503d71cf6e7d0d84b5a74f6c3d7bce9393\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0b789378e684003ad64e22874d77a4c6f160b9d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0b789378e684003ad64e22874d77a4c6f160b9d0"}]},"branch":"refs/heads/master"},"a74ecd71a746f28a69bf8b3fd883399d77281a9f":{"kind":"REWORK","_number":3,"created":"2019-05-10 16:26:24.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/51/658251/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/51/658251/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/3"}}},"commit":{"parents":[{"commit":"582a189117e861f8ceb73e8fe1e1d20dda083c59","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/582a189117e861f8ceb73e8fe1e1d20dda083c59"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 03:48:18.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 16:26:19.000000000","tz":-240},"subject":"[redhat] Install rsync-daemon for newer releases","message":"[redhat] Install rsync-daemon for newer releases\n\nThe newer packages havee moved rsyncd into its own package called\nrsync-daemon so we\u0027ll install that for anything newer than EL7.\n\nChange-Id: Id1e8cc503d71cf6e7d0d84b5a74f6c3d7bce9393\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a74ecd71a746f28a69bf8b3fd883399d77281a9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a74ecd71a746f28a69bf8b3fd883399d77281a9f"}]},"branch":"refs/heads/master"},"ba7b358ccd9dbe042ee6cdec0706abfecd7ce631":{"kind":"REWORK","_number":4,"created":"2019-05-10 16:56:27.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/51/658251/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/51/658251/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/51/658251/4"}}},"commit":{"parents":[{"commit":"582a189117e861f8ceb73e8fe1e1d20dda083c59","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/582a189117e861f8ceb73e8fe1e1d20dda083c59"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 03:48:18.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 16:56:22.000000000","tz":-240},"subject":"[redhat] Install rsync-daemon for newer releases","message":"[redhat] Install rsync-daemon for newer releases\n\nThe newer packages havee moved rsyncd into its own package called\nrsync-daemon so we\u0027ll install that for anything newer than EL7.\n\nChange-Id: Id1e8cc503d71cf6e7d0d84b5a74f6c3d7bce9393\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ba7b358ccd9dbe042ee6cdec0706abfecd7ce631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ba7b358ccd9dbe042ee6cdec0706abfecd7ce631"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
