)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~660788","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I6913341674dbde5524c2270e824bda4544211eca","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"sync-limited","hashtags":[],"change_id":"I6913341674dbde5524c2270e824bda4544211eca","subject":"Allow remote detection of repo sync status","status":"MERGED","created":"2019-05-22 17:16:56.000000000","updated":"2020-10-31 14:58:18.000000000","submitted":"2020-10-31 14:57:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660788-1604156234255-c7592bc5","meta_rev_id":"35383c03d1f61b448f19d527f79a57f1b8ff9dae","_number":660788,"virtual_id_number":660788,"owner":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"date":"2020-10-29 14:31:46.000000000","_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-31 14:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2020-10-27 13:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-30 14:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-30 14:32:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-03 10:01:02.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2020-10-27 13:09:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-30 14:32:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-31 14:57:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6f544e41906e63f463f2f3ebc101f13acb8f394","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2019-05-22 17:16:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7e37b4b3951505e730f66fb12e58278ec7931db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 20:10:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/88/660788/1/check/openstack-ansible-linters/3cd5de4/ : SUCCESS in 5m 44s\n- openstack-ansible-deploy-aio_metal-debian-stable finger://ze08.openstack.org/8f5b1e59ec734e70b704109f32585181 : RETRY_LIMIT in 26m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 finger://ze12.openstack.org/0013527811eb424e8d5f3760736859a7 : RETRY_LIMIT in 40m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-150 finger://ze11.openstack.org/9fcb633764924d3998e920dd2a4a34ae : RETRY_LIMIT in 43m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic finger://ze08.openstack.org/9ad9e81a123c426c8d7b5fdd0900090d : RETRY_LIMIT in 38m 02s\n- openstack-tox-docs http://logs.openstack.org/88/660788/1/check/openstack-tox-docs/a84b0ff/html/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d26babd01aa1bb2be5d52a22685c907b0bb5a6e9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-22 20:13:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4cdac99d8c47973214401a816cdc93ce8500dfef","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2019-05-23 09:45:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"890f2874d91a8eb8bb1a1c2912b41f644036f7cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 11:06:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/88/660788/1/check/openstack-ansible-linters/cc74097/ : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/88/660788/1/check/openstack-ansible-deploy-aio_metal-debian-stable/c530ff5/ : SUCCESS in 58m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/88/660788/1/check/openstack-ansible-deploy-aio_metal-centos-7/e6e5d2d/ : SUCCESS in 1h 13m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/88/660788/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/f9d5088/ : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/88/660788/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/fd18642/ : SUCCESS in 1h 01m 55s\n- openstack-tox-docs http://logs.openstack.org/88/660788/1/check/openstack-tox-docs/5e1da00/html/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"561c39da5f87fffa5da047460cbc182698c35b4f","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2019-05-23 11:22:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ba5944f806e7d4740d414b5703857c77ee1d9cd","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2019-05-23 11:27:34.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"dbac0f62c66bf0df56127bc6bd9deecb44825696","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 12:54:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/88/660788/3/check/openstack-ansible-linters/3c947a8/ : SUCCESS in 5m 52s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/88/660788/3/check/openstack-ansible-deploy-aio_metal-debian-stable/25af935/ : SUCCESS in 1h 01m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/88/660788/3/check/openstack-ansible-deploy-aio_metal-centos-7/5303c43/ : SUCCESS in 1h 20m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/88/660788/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/15415cb/ : SUCCESS in 1h 04m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/88/660788/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/1238222/ : SUCCESS in 55m 29s\n- openstack-tox-docs http://logs.openstack.org/88/660788/3/check/openstack-tox-docs/d64be2e/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":3},{"id":"e92b9eb80466b51ee64d262198ba89b06ed83bca","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-28 09:11:23.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"69e7927bfb1cc5af56ad6c03b1aae3ba48a9be1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 10:50:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/88/660788/4/check/openstack-ansible-linters/065de83/ : SUCCESS in 9m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/88/660788/4/check/openstack-ansible-deploy-aio_metal-debian-stable/a124dac/ : SUCCESS in 58m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/88/660788/4/check/openstack-ansible-deploy-aio_metal-centos-7/4dec0ce/ : SUCCESS in 1h 36m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/88/660788/4/check/openstack-ansible-deploy-aio_metal-opensuse-150/0242c11/ : SUCCESS in 1h 28m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/88/660788/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c48dd80/ : SUCCESS in 1h 22m 42s\n- openstack-tox-docs http://logs.openstack.org/88/660788/4/check/openstack-tox-docs/b3108f1/html/ : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":4},{"id":"88fbc3c5190c4ae71e121ed1d4ead9168e70adcc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-01 09:30:03.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nI like the idea, just one small comment/question","accounts_in_message":[],"_revision_number":4},{"id":"2928e19807f7ea4daba648f2de15e8060fed3d2e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-01 10:10:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5ac299140b894edaedd6d0ead931c3359f733123","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-01 15:50:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e92721ffbf15ae4f7e3abf303d16ec01a4984a8d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-01-10 21:56:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e8fbf520020ef2af0d60221b0019c0ba05fd3317","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-02-03 10:01:02.000000000","message":"Patch Set 4:\n\nWould you guys mind verifying my findings on relative paths in the exclude statement. https://review.opendev.org/#/c/705404/ I stumbled upon it while trying to exclude \"repo_build_global_links_path\" (/var/www/repo/links) dir, and only the relative path prevented files from getting synced over. See https://review.opendev.org/#/c/705390/","accounts_in_message":[],"_revision_number":4},{"id":"edfa7e6ec98c648d9ea6065a9d156e91bc8c466b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-03 13:43:36.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"5495f1b813c812774372f9ee5680e469a5005113","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-27 11:40:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"29eb3fd9f6fc8bd7a11e0d60cb69f341da77f738","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-27 13:09:41.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d7e8c1af1513ec882d6350e7f7fd38d386bf5753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 13:41:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1841beaa36a64263bd7c168b37bcfb41 : TIMED_OUT in 30m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b4329934973245c18b3b12fb5d1f8338 : SUCCESS in 1h 07m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d294fba913c64fc69c54a3a6246a5844 : SUCCESS in 1h 25m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb59524e48334c59912512e7ba82b2fa : SUCCESS in 1h 15m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40ddf86defd6454980e0d023d54bd7b8 : SUCCESS in 1h 07m 24s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d81f14bcb3fa47b686c8dd722ccf248c : SUCCESS in 1h 53m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80c41233b095467595c3eb5158707308 : SUCCESS in 1h 30m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097ae6c25814422ea8aff5606e193d84 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":5},{"id":"85bae445f7c6380930f219173267661ec688dfba","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-10-29 14:31:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"478789223d104d0bc527e9f9925b6224ceb5293f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 16:47:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8dce673dac7e4aeaa238045d1da61dec : SUCCESS in 11m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cf33e0fa9a9742269e4b476bda1ef681 : FAILURE in 59m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ce967f5b7c7c41debe330ca8177660af : FAILURE in 22m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c77ce83a3316475dbbe2e560daee85cf : SUCCESS in 1h 15m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1cdf9f29ff0437a8e7756f30790ee40 : SUCCESS in 1h 12m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/53cf83d40e954c7f855e6071c19cd7c7 : SUCCESS in 2h 04m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b234d045526455ba2659f2c7790b7ae : SUCCESS in 1h 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/122554527c2444ebb1fa317f56260f73 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":5},{"id":"5f5e1f779f2aec2cf164f018771c89a316d10e29","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-29 19:37:55.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"97a901ce78489889763f9a6ec3fc3b7b3f777e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 21:44:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/713019360116453db2905d297ee0f1e8 : SUCCESS in 8m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/aa614f81ab1548fca817f5720dcb959c : SUCCESS in 1h 12m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a83673edc4874833b706cf8b43471a14 : SUCCESS in 1h 33m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b629fe743ea4f1697ca5e4ded360366 : SUCCESS in 1h 39m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d638cdc3ac448f2bc54f7e36bc22274 : SUCCESS in 1h 05m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/470f6e3b8acf4d998fc532fd17e5d90d : SUCCESS in 1h 46m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fd2ea9b54084872b889136d734cfbf5 : SUCCESS in 1h 42m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f9d4c5e3cd4563834c18cb2707de2c : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":5},{"id":"3c5edae953580b1cff09b23c438a1aa51b5eee28","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-30 14:32:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e1ff581f69d9faccdf40d85b6b5363f5827120f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 14:32:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1490f18922ba52374b4e7aff5b27b11d6396f93c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 16:06:14.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/215efc98812e4b87a0fb6c0afd788ba2 : SUCCESS in 14m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a5b19a7571a94ab1a5b984b0d58569de : FAILURE in 34m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a29d6ba2fd1b432d984fc682165ebd4a : SUCCESS in 1h 24m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c187ce8203f9454198c7e8d86d3e1002 : SUCCESS in 1h 24m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/225ee00fe17041aa842ec69f2f9384b6 : SUCCESS in 1h 16m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9268d893bf94482a87a981cdf70e9d0 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":5},{"id":"dba881dc3310d4d47eb3cb946016e7a5aff906b3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-31 10:21:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1a494a56efe733f4f9567ee95d4b81f76cc5e564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 12:39:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ab3032ccb89d4925b9e8664c8c8b77e8 : SUCCESS in 7m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d29b076aeab84f7d80931db7d5c91e1a : SUCCESS in 1h 10m 05s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4e2bb7b234f74a1eb505bf6f96df276e : SUCCESS in 1h 31m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf1e11b03f5247aba20d702ca9651ecf : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f32925b63c0249ab820ad7d7d2e9facf : SUCCESS in 1h 14m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cc046e03ce2a4ee28973a697748a9d4f : SUCCESS in 2h 14m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12f3d685c62444789e41c60c2acc4b08 : SUCCESS in 1h 21m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c8fb06dbd64a3f9dbe528e59683979 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":5},{"id":"d1840f4dc2b84245e9895fe00090c71ed07856e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 12:39:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"895e9db4845cd352237728128dfa571598d3daf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 14:57:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/41bf3544670a4a4db46ab634077c59b3 : SUCCESS in 7m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d80182efe578463a8d1260dc4caa5289 : SUCCESS in 1h 56m 08s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/583ac0bf3a1443d38e845e37e359974d : SUCCESS in 1h 24m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7bfe101d2b6481bbb5b47eec87be384 : SUCCESS in 1h 12m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a124f260aed248fcbfca1bb8f780cde4 : SUCCESS in 1h 06m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938773291a014eada442f50f6d5df6ee : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":5},{"id":"ca26a0699569b84183005eb3cf5bcd3f7aca8794","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 14:57:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"c9c14c0d776a8a87cb5c9716f757300fc7679f84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 14:58:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c616a5922e5496ca66c866ff999fbba : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7132acbd3b82fd22f3ff88f313b12f1e77a76c24","revisions":{"b32415098c6570a634a60c4642ed209a48563190":{"kind":"REWORK","_number":1,"created":"2019-05-22 17:16:56.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/88/660788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/88/660788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/1"}}},"commit":{"parents":[{"commit":"3a2ff676efe501df9f5019270253b82acd2536e1","subject":"Clarify supported SUSE versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3a2ff676efe501df9f5019270253b82acd2536e1"}]}],"author":{"name":"Georgina","email":"georgina.shippey@bbc.co.uk","date":"2019-05-22 17:05:01.000000000","tz":0},"committer":{"name":"Georgina","email":"georgina.shippey@bbc.co.uk","date":"2019-05-22 17:16:49.000000000","tz":0},"subject":"Improving repo sync and HAProxy checks","message":"Improving repo sync and HAProxy checks\n\nIf repo-install is run limited to a single container then the\ncontainer will not be synced as sync only starts when changes are\nmade to the master repo server, or when lsyncd is restarted.\n- Changes made to delegate a restart of lsyncd to the master when it\n is not already included in the limit. Better sync ensurance.\n\nUsing a naive method so HAProxy doesnt delegate to non synchronised\nrepo server containers.\n- file called repo-sync-complete needs to be in /var/www/repo for\n  HAProxy to be satisfied.\n- file is automatically provisioned to the master repo server\n- file is provisioned to slaves in the repo_pre_post script.\n\nChange-Id: I6913341674dbde5524c2270e824bda4544211eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b32415098c6570a634a60c4642ed209a48563190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b32415098c6570a634a60c4642ed209a48563190"}]},"branch":"refs/heads/master"},"097f69c6a1f7f96270d0131e6c7e9feb2692d30a":{"kind":"REWORK","_number":2,"created":"2019-05-23 11:22:54.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/88/660788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/88/660788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/2"}}},"commit":{"parents":[{"commit":"3a2ff676efe501df9f5019270253b82acd2536e1","subject":"Clarify supported SUSE versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3a2ff676efe501df9f5019270253b82acd2536e1"}]}],"author":{"name":"Georgina","email":"georgina.shippey@bbc.co.uk","date":"2019-05-22 17:05:01.000000000","tz":0},"committer":{"name":"Georgina","email":"georgina.shippey@bbc.co.uk","date":"2019-05-23 11:22:36.000000000","tz":0},"subject":"Improving repo sync and HAProxy checks","message":"Improving repo sync and HAProxy checks\n\nIf repo-install is run limited to a single container then the\ncontainer will not be synced as sync only starts when changes are\nmade to the master repo server, or when lsyncd is restarted.\n- Changes made to delegate a restart of lsyncd to the master when it\n is not already included in the limit. Better sync ensurance.\n\nUsing a naive method so HAProxy doesnt delegate to non synchronised\nrepo server containers.\n- file called repo-sync-complete needs to be in /var/www/repo for\n  HAProxy to be satisfied.\n- file is automatically provisioned to the master repo server\n- file is provisioned to slaves in the repo_pre_post script.\n\nChange-Id: I6913341674dbde5524c2270e824bda4544211eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/097f69c6a1f7f96270d0131e6c7e9feb2692d30a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/097f69c6a1f7f96270d0131e6c7e9feb2692d30a"}]},"branch":"refs/heads/master"},"59b28871f4ba667be93ffbdc065efdc8f8c0fb7f":{"kind":"REWORK","_number":3,"created":"2019-05-23 11:27:34.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/88/660788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/88/660788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/3"}}},"commit":{"parents":[{"commit":"3a2ff676efe501df9f5019270253b82acd2536e1","subject":"Clarify supported SUSE versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3a2ff676efe501df9f5019270253b82acd2536e1"}]}],"author":{"name":"Georgina","email":"georgina.shippey@bbc.co.uk","date":"2019-05-22 17:05:01.000000000","tz":0},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2019-05-23 11:27:30.000000000","tz":0},"subject":"Improving repo sync and HAProxy checks","message":"Improving repo sync and HAProxy checks\n\nIf repo-install is run limited to a single container then the\ncontainer will not be synced as sync only starts when changes are\nmade to the master repo server, or when lsyncd is restarted.\n- Changes made to delegate a restart of lsyncd to the master when it\n is not already included in the limit. Better sync ensurance.\n\nUsing a naive method so HAProxy doesnt delegate to non synchronised\nrepo server containers.\n- file called repo-sync-complete needs to be in /var/www/repo for\n  HAProxy to be satisfied.\n- file is automatically provisioned to the master repo server\n- file is provisioned to slaves in the repo_pre_post script.\n\nChange-Id: I6913341674dbde5524c2270e824bda4544211eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/59b28871f4ba667be93ffbdc065efdc8f8c0fb7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/59b28871f4ba667be93ffbdc065efdc8f8c0fb7f"}]},"branch":"refs/heads/master"},"e5cff672fae7bc111ae3be7b0492344e7264853e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-05-28 09:11:23.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/88/660788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/88/660788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/4"}}},"commit":{"parents":[{"commit":"3a2ff676efe501df9f5019270253b82acd2536e1","subject":"Clarify supported SUSE versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3a2ff676efe501df9f5019270253b82acd2536e1"}]}],"author":{"name":"Georgina","email":"georgina.shippey@bbc.co.uk","date":"2019-05-22 17:05:01.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-05-28 09:10:34.000000000","tz":0},"subject":"Allow remote detection of repo sync status","message":"Allow remote detection of repo sync status\n\nIf a repo container and its data are deleted and recreated then it is\nnot currently possible for a loadbalancer healthcheck to differentiate\nbetween an empty repo server and a correctly synchronised one.\n\nThis patch creates a file \u0027repo-sync-complete\u0027 as part of the process\nof synchronising repo contents from master repo servers to slaves. The\npresence of this file on the slave can then be used as the loadbalancer\nhealthcheck to ensure that repo contents are only served once sync has\ncompleted.\n\nIn addition, this patch ensures that synchronisation occurs from the\nmaster to a reprovisioned slave by triggering a master repo server lsyncd\nrestart handler during the initial setup of the slave repo server.\nCurrently, a freshly provisioned repo server will remain empty\nfor an indeterminate amount of time, this patch forces a complete re-sync\nto occur.\n\nChange-Id: I6913341674dbde5524c2270e824bda4544211eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e5cff672fae7bc111ae3be7b0492344e7264853e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e5cff672fae7bc111ae3be7b0492344e7264853e"}]},"branch":"refs/heads/master"},"7132acbd3b82fd22f3ff88f313b12f1e77a76c24":{"kind":"REWORK","_number":5,"created":"2020-10-27 11:40:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/660788/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/88/660788/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/88/660788/5"}}},"commit":{"parents":[{"commit":"7f58e1cc61f15a2edf5f6e6078cdbead4f504c87","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7f58e1cc61f15a2edf5f6e6078cdbead4f504c87"}]}],"author":{"name":"Georgina","email":"georgina.shippey@bbc.co.uk","date":"2019-05-22 17:05:01.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-27 11:37:21.000000000","tz":120},"subject":"Allow remote detection of repo sync status","message":"Allow remote detection of repo sync status\n\nIf a repo container and its data are deleted and recreated then it is\nnot currently possible for a loadbalancer healthcheck to differentiate\nbetween an empty repo server and a correctly synchronised one.\n\nThis patch creates a file \u0027repo-sync-complete\u0027 as part of the process\nof synchronising repo contents from master repo servers to slaves. The\npresence of this file on the slave can then be used as the loadbalancer\nhealthcheck to ensure that repo contents are only served once sync has\ncompleted.\n\nIn addition, this patch ensures that synchronisation occurs from the\nmaster to a reprovisioned slave by triggering a master repo server lsyncd\nrestart handler during the initial setup of the slave repo server.\nCurrently, a freshly provisioned repo server will remain empty\nfor an indeterminate amount of time, this patch forces a complete re-sync\nto occur.\n\nChange-Id: I6913341674dbde5524c2270e824bda4544211eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7132acbd3b82fd22f3ff88f313b12f1e77a76c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7132acbd3b82fd22f3ff88f313b12f1e77a76c24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
