)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~663585","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I1bbce6b58f55f0715b3005082d891f301ce4b0b3","project":"openstack/openstack-ansible-repo_server","branch":"master","hashtags":[],"change_id":"I1bbce6b58f55f0715b3005082d891f301ce4b0b3","subject":"Cap sphinx for py2 to match global requirements","status":"MERGED","created":"2019-06-06 10:16:24.000000000","updated":"2019-07-10 18:51:14.000000000","submitted":"2019-07-10 18:51:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663585-1562784674353-a65eadfb","meta_rev_id":"65ad284f8cc68d2a8253cb6c693de965cdb84902","_number":663585,"virtual_id_number":663585,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-10 18:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-10 18:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-10 18:37:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 18:37:58.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-10 18:51:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5303c59b59299498cde3b8260719d45ad234bee","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-06-06 10:16:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96906cafd02df2941e885afd60a06a663329d804","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:24:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/85/663585/1/check/requirements-check/fe0f1c0/ : SUCCESS in 1m 58s\n- openstack-ansible-linters http://logs.openstack.org/85/663585/1/check/openstack-ansible-linters/ca53a46/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/85/663585/1/check/openstack-tox-docs/317615c/html/ : SUCCESS in 2m 19s\n- build-openstack-releasenotes http://logs.openstack.org/85/663585/1/check/build-openstack-releasenotes/fb52007/html/ : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":1},{"id":"316030b0b681f791868aba22bf40cffef9ea4cf5","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-10 18:37:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88e8fa72871ac4a34511fb2a57dc4a8f049ed822","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:38:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29234fd2ad539ef6e9f474ba1de922ce294c991e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:51:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"65ad284f8cc68d2a8253cb6c693de965cdb84902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:51:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/85/663585/1/gate/requirements-check/fa22ed8/ : SUCCESS in 2m 21s\n- openstack-ansible-linters http://logs.openstack.org/85/663585/1/gate/openstack-ansible-linters/6209eb1/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/85/663585/1/gate/openstack-tox-docs/b827b10/html/ : SUCCESS in 2m 49s\n- build-openstack-releasenotes http://logs.openstack.org/85/663585/1/gate/build-openstack-releasenotes/e574fa9/html/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"521c50388790f909c2f3a90b764852eb2945f26b","revisions":{"521c50388790f909c2f3a90b764852eb2945f26b":{"kind":"REWORK","_number":1,"created":"2019-06-06 10:16:24.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/85/663585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/85/663585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/85/663585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/85/663585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/85/663585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/85/663585/1"}}},"commit":{"parents":[{"commit":"3a2ff676efe501df9f5019270253b82acd2536e1","subject":"Clarify supported SUSE versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3a2ff676efe501df9f5019270253b82acd2536e1"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-06 10:16:04.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-06 10:16:19.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: I1bbce6b58f55f0715b3005082d891f301ce4b0b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/521c50388790f909c2f3a90b764852eb2945f26b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/521c50388790f909c2f3a90b764852eb2945f26b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
