)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~664115","triplet_id":"openstack%2Fopenstack-ansible-repo_server~stable%2Fqueens~If1a8162b45dc50e65a9f60431a0ab7283f59e7e0","project":"openstack/openstack-ansible-repo_server","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"If1a8162b45dc50e65a9f60431a0ab7283f59e7e0","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-07 23:30:59.000000000","updated":"2019-06-11 22:22:03.000000000","submitted":"2019-06-11 22:22:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664115-1560291723050-d9d96933","meta_rev_id":"42cb30c63da79a32daad4b1651a676e1f12c529f","_number":664115,"virtual_id_number":664115,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-11 22:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 21:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 21:55:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 21:55:30.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 22:22:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19e95c4ddd228c9410762786466b81c369140164","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:30:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd816aed14a66d17df443e98e736d997ae3f1e91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:04:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/15/664115/1/check/build-openstack-sphinx-docs/a8b92d3/html/ : SUCCESS in 3m 43s\n- openstack-ansible-linters http://logs.openstack.org/15/664115/1/check/openstack-ansible-linters/13511f1/ : SUCCESS in 5m 26s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/15/664115/1/check/openstack-ansible-functional-centos-7/ec23056/ : SUCCESS in 21m 44s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/15/664115/1/check/openstack-ansible-functional-opensuse-423/ffc2c52/ : FAILURE in 9m 07s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/15/664115/1/check/openstack-ansible-functional-ubuntu-xenial/922dfb9/ : SUCCESS in 11m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6f7779c0d91686592dbc17b2e47ba77c8e200b27","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 20:55:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"65c3afc53921147c2971717c323baae1d62aff70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:35:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/15/664115/1/check/build-openstack-sphinx-docs/218c376/html/ : SUCCESS in 3m 32s\n- openstack-ansible-linters http://logs.openstack.org/15/664115/1/check/openstack-ansible-linters/012bfef/ : SUCCESS in 6m 08s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/15/664115/1/check/openstack-ansible-functional-centos-7/a02657b/ : SUCCESS in 20m 03s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/15/664115/1/check/openstack-ansible-functional-ubuntu-xenial/f546c76/ : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fdfd3e3208d0fa2a6ef240ba90fce6d8258fa4db","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 21:55:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f293b115c0c8b12ca78ae608f3f9e4ee738c1941","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:55:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfa2f429f6242dd257145b2939e87d2bac7e734e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:22:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/15/664115/1/gate/build-openstack-sphinx-docs/93d6fa2/html/ : SUCCESS in 3m 46s\n- openstack-ansible-linters http://logs.openstack.org/15/664115/1/gate/openstack-ansible-linters/73cbbf5/ : SUCCESS in 4m 46s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/15/664115/1/gate/openstack-ansible-functional-centos-7/80e3b51/ : SUCCESS in 19m 59s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/15/664115/1/gate/openstack-ansible-functional-ubuntu-xenial/75cb0e7/ : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8166856d42e548e5b92a3d4a6e8965488a7c72f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:22:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea151abb64cb1eeca2e3e3c4905e1429b4e20295","revisions":{"ea151abb64cb1eeca2e3e3c4905e1429b4e20295":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:30:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/664115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/15/664115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/15/664115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/15/664115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/15/664115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/15/664115/1"}}},"commit":{"parents":[{"commit":"79661b7d04a763cd2cef1933453625b3ee793abf","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/79661b7d04a763cd2cef1933453625b3ee793abf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:30:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:30:55.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: If1a8162b45dc50e65a9f60431a0ab7283f59e7e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ea151abb64cb1eeca2e3e3c4905e1429b4e20295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ea151abb64cb1eeca2e3e3c4905e1429b4e20295"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
