)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~664361","triplet_id":"openstack%2Fopenstack-ansible-repo_server~stable%2Frocky~Ic92d1db65c97153bae4fec1cae90db865a6beecf","project":"openstack/openstack-ansible-repo_server","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ic92d1db65c97153bae4fec1cae90db865a6beecf","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:45:45.000000000","updated":"2019-06-10 17:50:30.000000000","submitted":"2019-06-10 17:50:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664361-1560189030584-7fce86ad","meta_rev_id":"bcc89aaa3e5a7bdac0403da9abd3e9a69e145725","_number":664361,"virtual_id_number":664361,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 17:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:19:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:19:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:19:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 17:50:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa7dae4f5eccdd08ff7c9061b870556658a0c1b8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:45:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb8a81d34832b9ca715ed65069081cf2e918723d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:38:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/61/664361/1/check/openstack-ansible-linters/6e58235/ : SUCCESS in 4m 59s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/61/664361/1/check/openstack-ansible-functional-centos-7/57795c8/ : SUCCESS in 21m 46s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/61/664361/1/check/openstack-ansible-functional-opensuse-423/08b6522/ : FAILURE in 7m 52s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/61/664361/1/check/openstack-ansible-functional-opensuse-150/f922892/ : FAILURE in 10m 33s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/61/664361/1/check/openstack-ansible-functional-ubuntu-bionic/157d65c/ : SUCCESS in 13m 07s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/61/664361/1/check/openstack-ansible-functional-ubuntu-xenial/8b132ab/ : SUCCESS in 13m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/664361/1/check/build-openstack-sphinx-docs/0469ebf/html/ : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8a976ef102f16ed45600db4bccf5e1fe14d8d60d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:19:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aba4f348d9c49b09afe98bc96a13be67104cda8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:19:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd548f14bf6f655be8ae0b0a5fc82099e751a5e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:50:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bdad587e7f6b01b45350b2214779a534180fa2a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:50:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/61/664361/1/gate/openstack-ansible-linters/8e6ca93/ : SUCCESS in 6m 37s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/61/664361/1/gate/openstack-ansible-functional-centos-7/39c6388/ : SUCCESS in 23m 27s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/61/664361/1/gate/openstack-ansible-functional-opensuse-423/5dbcc16/ : FAILURE in 12m 44s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/61/664361/1/gate/openstack-ansible-functional-ubuntu-bionic/54a352c/ : SUCCESS in 12m 53s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/61/664361/1/gate/openstack-ansible-functional-ubuntu-xenial/a9f0226/ : SUCCESS in 15m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/664361/1/gate/build-openstack-sphinx-docs/5711743/html/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"277b4140174f7fd45c06e3230a0e4a3780c28c67","revisions":{"277b4140174f7fd45c06e3230a0e4a3780c28c67":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:45:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/664361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/61/664361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/61/664361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/61/664361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/61/664361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/61/664361/1"}}},"commit":{"parents":[{"commit":"561b90bb984eb8e0f52a1f6bfa67f9213f8ebf68","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/561b90bb984eb8e0f52a1f6bfa67f9213f8ebf68"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:44.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ic92d1db65c97153bae4fec1cae90db865a6beecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/277b4140174f7fd45c06e3230a0e4a3780c28c67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/277b4140174f7fd45c06e3230a0e4a3780c28c67"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
