)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~666368","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I65fce322651bf67812449143525063b13902b455","project":"openstack/openstack-ansible-repo_server","branch":"master","hashtags":[],"change_id":"I65fce322651bf67812449143525063b13902b455","subject":"Sync Sphinx requirement","status":"ABANDONED","created":"2019-06-19 16:00:51.000000000","updated":"2019-07-18 13:30:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"024287a40c8deee49af0ae69b3819e0d4fdec23b","_number":666368,"virtual_id_number":666368,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-19 16:23:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 16:23:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9ab4c9c59c66ffe3d3db2ad86d6de31490609a4","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-19 16:00:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"019d5fd541751f444482e32f6c455037ee7ab04d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:23:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/68/666368/1/check/requirements-check/0a2302c/ : SUCCESS in 2m 12s\n- openstack-ansible-linters http://logs.openstack.org/68/666368/1/check/openstack-ansible-linters/ead97d3/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/68/666368/1/check/openstack-tox-docs/2e7aad0/html/ : SUCCESS in 3m 49s\n- build-openstack-releasenotes http://logs.openstack.org/68/666368/1/check/build-openstack-releasenotes/0474fec/html/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"024287a40c8deee49af0ae69b3819e0d4fdec23b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-18 13:30:50.000000000","message":"Abandoned\n\nalready merged in other patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50d6f524d3c5b5df279847d28977d8b93ce7fd26","revisions":{"50d6f524d3c5b5df279847d28977d8b93ce7fd26":{"kind":"REWORK","_number":1,"created":"2019-06-19 16:00:51.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/68/666368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/68/666368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/68/666368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/68/666368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/68/666368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/68/666368/1"}}},"commit":{"parents":[{"commit":"3a2ff676efe501df9f5019270253b82acd2536e1","subject":"Clarify supported SUSE versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3a2ff676efe501df9f5019270253b82acd2536e1"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-19 16:00:40.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-19 16:00:47.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\nChange-Id: I65fce322651bf67812449143525063b13902b455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/50d6f524d3c5b5df279847d28977d8b93ce7fd26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/50d6f524d3c5b5df279847d28977d8b93ce7fd26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
