)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~671627","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I916c9d8d4bc4772083e17b9b9fec598194a230fe","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I916c9d8d4bc4772083e17b9b9fec598194a230fe","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-18 22:20:56.000000000","updated":"2019-07-19 13:47:08.000000000","submitted":"2019-07-19 13:47:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671627-1563544028346-6d7173bf","meta_rev_id":"e3a7cedc95f033c1e79bbc86447a74235e9f0061","_number":671627,"virtual_id_number":671627,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 13:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-19 12:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-19 12:08:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 12:08:40.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 13:47:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"983839b4e201174fefc68bfa5d85a9980ece7db4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-18 22:20:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0316ad408c7e89038631a76eb6fe8226ca4ace3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 00:02:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/27/671627/1/check/openstack-ansible-linters/2b9c0d5/ : SUCCESS in 5m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/27/671627/1/check/openstack-ansible-deploy-aio_metal-debian-stable/2fabcf5/ : SUCCESS in 1h 05m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/27/671627/1/check/openstack-ansible-deploy-aio_metal-centos-7/6791704/ : SUCCESS in 1h 28m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/27/671627/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/8d2c200/ : SUCCESS in 1h 13m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/27/671627/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2873a14/ : SUCCESS in 51m 03s\n- openstack-tox-docs http://logs.openstack.org/27/671627/1/check/openstack-tox-docs/1110920/html/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c80e9dca4e73599e8a8070830f23bb0442b72971","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 12:08:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37943ec0690418832b112f5cad95c29004082197","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:08:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d32bd9db0b4c80f11a5fa6bad93a12aa2270afd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 13:47:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e5c00c951231578eb686d5a733c8494f39b521d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 13:47:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/27/671627/1/gate/openstack-ansible-linters/09f06a9/ : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/27/671627/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/25332a4/ : SUCCESS in 1h 06m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/27/671627/1/gate/openstack-ansible-deploy-aio_metal-centos-7/225478e/ : SUCCESS in 1h 33m 10s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/27/671627/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/045a776/ : SUCCESS in 1h 14m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/27/671627/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/fe058a2/ : SUCCESS in 1h 04m 28s\n- openstack-tox-docs http://logs.openstack.org/27/671627/1/gate/openstack-tox-docs/a832d76/html/ : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9ab6c4bd2e786730726a7d9d3edf9c7c56f23b0","revisions":{"c9ab6c4bd2e786730726a7d9d3edf9c7c56f23b0":{"kind":"REWORK","_number":1,"created":"2019-07-18 22:20:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/671627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/27/671627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/27/671627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/27/671627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/27/671627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/27/671627/1"}}},"commit":{"parents":[{"commit":"8c7151fe00ac509642919f39f1264968f14fa732","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/8c7151fe00ac509642919f39f1264968f14fa732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:20:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:20:54.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I916c9d8d4bc4772083e17b9b9fec598194a230fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c9ab6c4bd2e786730726a7d9d3edf9c7c56f23b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c9ab6c4bd2e786730726a7d9d3edf9c7c56f23b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
